)]}'
{"id":"openstack%2Faodh~920738","triplet_id":"openstack%2Faodh~master~Ia26371b6332f858e7ade81556c93aa26b172230b","project":"openstack/aodh","branch":"master","topic":"zuul-test","attention_set":{},"removed_from_attention_set":{"5202":{"account":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"last_update":"2024-06-04 09:07:29.000000000","reason":"\u003cGERRIT_ACCOUNT_5202\u003e replied on the change","reason_account":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"}}},"hashtags":[],"change_id":"Ia26371b6332f858e7ade81556c93aa26b172230b","subject":"[DNM] Testing zuul config","status":"ABANDONED","created":"2024-05-29 13:03:35.000000000","updated":"2024-06-04 10:58:39.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"39974d76ba6580647e943ec14c03f1a7d95abbb7","_number":920738,"virtual_id_number":920738,"owner":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2024-06-04 10:23:10.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-29 13:04:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c9dbd8c3ac7a7faff6f75fbd28ccce0c83d9f6cf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2024-05-29 13:03:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a154d7c508db492c3ecf4814bb7348799625dd7d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-29 13:04:06.000000000","message":"Patch Set 1: Verified-1\n\n(3 comments)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/aodh on branch master.  The\nproblem was:\n\n  Job telemetry-tempest-base not defined\n\nThe problem appears in the the \"aodh-tempest-plugin-postgresql\" job stanza:\n\n  job:\n      name: aodh-tempest-plugin-postgresql\n      parent: telemetry-tempest-base\n      irrelevant-files: *aodh-irrelevant-files\n      vars:\n        devstack_services:\n          postgresql: true\n          mysql: false\n        devstack_localrc:\n  ...\n\n  in \"openstack/aodh/.zuul.yaml@master\", line 59","accounts_in_message":[],"_revision_number":1},{"id":"eadebec2a7a0cb39dc9b47459a1ee52c0804b8b8","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2024-05-31 10:09:27.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"eae21a6af3626ab768980475bfa4228f1f5a7217","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-31 10:09:49.000000000","message":"Patch Set 1:\n\n(3 comments)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/aodh on branch master.  The\nproblem was:\n\n  Job telemetry-tempest-base not defined\n\nThe problem appears in the the \"aodh-tempest-plugin-postgresql\" job stanza:\n\n  job:\n      name: aodh-tempest-plugin-postgresql\n      parent: telemetry-tempest-base\n      irrelevant-files: *aodh-irrelevant-files\n      vars:\n        devstack_services:\n          postgresql: true\n          mysql: false\n        devstack_localrc:\n  ...\n\n  in \"openstack/aodh/.zuul.yaml@master\", line 59","accounts_in_message":[],"_revision_number":1},{"id":"4746459e74df5571042b10b57809b7b2830e589c","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2024-06-04 09:07:29.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6b93215907e69ab3b28d39c00f32f26a703ea778","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-04 10:23:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cb60d06758c24c20bd8e949ee1fdd4a3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa841ceaeeb0443685a1e76ab0646de6 : SUCCESS in 3m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/470b8ca6b76b48dd8e07f972c934269e : SUCCESS in 5m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7e7a895b190947b2be4d1cd8e3bbeff2 : SUCCESS in 4m 59s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0df75f1d02b745cd95e37069e7ffe187 : SUCCESS in 8m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb33269f9b3448fa81643e0b88ce530c : SUCCESS in 7m 02s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/d309cb3e2cf145aa90316c45f7ab0bf4 : SUCCESS in 46m 53s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/b93e8fa0ea034c45807d50c991a3cfcc : SUCCESS in 43m 32s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/c8f0d90a431d4eb59fe9bbc66f65f556 : SUCCESS in 40m 41s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/6a2c1f97b4c24ff8953bea47471cf6c7 : SUCCESS in 38m 46s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/220a102b31dc49d1a100f4dbf5a37fd9 : SUCCESS in 1h 02m 09s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/c1fc6b4fe0fa409aa3fc1405ca959862 : SUCCESS in 1h 02m 39s","accounts_in_message":[],"_revision_number":1},{"id":"39974d76ba6580647e943ec14c03f1a7d95abbb7","tag":"autogenerated:gerrit:abandon","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2024-06-04 10:58:39.000000000","message":"Abandoned\n\nGate issue solved","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bd4ce6ac484b380fc3b739e29c908e3bbdcd7c3b","revisions":{"bd4ce6ac484b380fc3b739e29c908e3bbdcd7c3b":{"kind":"REWORK","_number":1,"created":"2024-05-29 13:03:35.000000000","uploader":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"ref":"refs/changes/38/920738/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/38/920738/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/38/920738/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/38/920738/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/38/920738/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/38/920738/1"}}},"commit":{"parents":[{"commit":"2a98c261f093aa084a499f1935fccb0cb2eee606","subject":"Merge \"Fix db upgrade with SQLAlchemy 2.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/2a98c261f093aa084a499f1935fccb0cb2eee606"}]}],"author":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2024-05-29 13:03:29.000000000","tz":60},"committer":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2024-05-29 13:03:29.000000000","tz":60},"subject":"[DNM] Testing zuul config","message":"[DNM] Testing zuul config\n\nChange-Id: Ia26371b6332f858e7ade81556c93aa26b172230b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/bd4ce6ac484b380fc3b739e29c908e3bbdcd7c3b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/bd4ce6ac484b380fc3b739e29c908e3bbdcd7c3b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
