)]}'
{"id":"openstack%2Faodh~925529","triplet_id":"openstack%2Faodh~master~Ifac31d1fc11b44fd6d3d18041f8bacee8287bccf","project":"openstack/aodh","branch":"master","topic":"secure-rbac","attention_set":{},"removed_from_attention_set":{"5202":{"account":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"last_update":"2024-08-26 10:22:37.000000000","reason":"\u003cGERRIT_ACCOUNT_5202\u003e replied on the change","reason_account":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"}},"4264":{"account":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"last_update":"2024-08-27 05:44:16.000000000","reason":"removed on reply"},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2024-08-27 06:38:25.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ifac31d1fc11b44fd6d3d18041f8bacee8287bccf","subject":"Keep new RBAC disable by default","status":"MERGED","created":"2024-08-01 19:27:00.000000000","updated":"2024-08-27 06:39:31.000000000","submitted":"2024-08-27 06:38:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"925529-secure-rbac","meta_rev_id":"a2f0a0c42eb8926d41a88d72250a4330ebacadd6","_number":925529,"virtual_id_number":925529,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-08-27 06:38:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"all":[{"value":2,"date":"2024-08-26 10:22:37.000000000","_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":2,"date":"2024-08-27 05:44:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"all":[{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":1,"date":"2024-08-27 05:44:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"}],"REVIEWER":[{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-01 20:11:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-08-02 07:19:47.000000000","updated_by":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"reviewer":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"state":"CC"},{"updated":"2024-08-21 00:20:51.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"},{"updated":"2024-08-21 00:20:51.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"state":"REVIEWER"}],"messages":[{"id":"bb064d7c17c90642502221555787c66eb808580a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-08-01 19:27:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"47f99b6d5e1b22015f0c7e6535f1a17b5721717a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-08-01 19:32:04.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"837273f02c27af4b52bf32aea4d62d6c9172ce06","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-08-01 19:33:33.000000000","message":"Topic srbac removed","accounts_in_message":[],"_revision_number":2},{"id":"6497fa9197d186127c80ceccbef6e83f9e33c82a","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-08-01 19:33:36.000000000","message":"Topic set to secure-rbac","accounts_in_message":[],"_revision_number":2},{"id":"35b1c4c06d63fb558394d24fed627234c2700178","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-01 20:11:23.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/39de088aa1094db7b21146adfa17c26b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/be1bb85110ef4c679a012576059ab48c : SUCCESS in 5m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d9c14ad2ad074c67ac283bf48c7407b1 : SUCCESS in 5m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8c071c09f9ce428ea9c5c1f7f9537a38 : SUCCESS in 5m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a4fb750077d54f29b515eb8d66fc7735 : SUCCESS in 11m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6e657aca69b442b9468a5d867d4f33d : SUCCESS in 8m 45s\n- requirements-check https://zuul.opendev.org/t/openstack/build/2003352c356d411e9e8593de5f9cd791 : SUCCESS in 2m 43s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/11cba079418146ad98ba13431c249666 : SUCCESS in 31m 09s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/6162e31a99994904a188253d7502bf87 : SUCCESS in 25m 26s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/45bc8b833fd644c198b600dc4f119fc3 : SUCCESS in 34m 33s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/839e0aca6591444c9d32cab0c932110b : SUCCESS in 38m 15s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/552b936ca0cd43b5a8fabc7118eba3a9 : FAILURE in 7m 58s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/f169f74dd8714c20bc71cf1fb4e5dedb : FAILURE in 9m 32s","accounts_in_message":[],"_revision_number":2},{"id":"73a0b8066ef0df539027da354f5d8ee2b272c1ae","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-08-01 23:48:50.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6dde3893109bb65705021cadf9cae279b2c5a598","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-02 00:51:51.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8ea1193d2e3f43a297b413cd18a60682\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4594b12726a5432e841a94f50b518a50 : SUCCESS in 3m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/27ddfab831c4461d8902ba6e7147d257 : SUCCESS in 7m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/03af66fcfae1442e830ffde793424e52 : SUCCESS in 4m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/706ff1b0995f4bc6b51b6d0487ab4474 : SUCCESS in 9m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe38e898fd624885856894a87d9d8aa9 : SUCCESS in 5m 51s\n- requirements-check https://zuul.opendev.org/t/openstack/build/3297124f93f14f0ca02a97d01284cdf4 : SUCCESS in 2m 34s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/ae41d93291584499b2d7e590077978d5 : SUCCESS in 32m 12s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/e36b5c5751264bac9cc2bd3f9818b387 : SUCCESS in 31m 52s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/2161ec077c4b4ab0ac9c96aa907d6bb2 : SUCCESS in 40m 02s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/df3823dbe65b4c16841035d42a550127 : SUCCESS in 38m 30s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/73cdba09b68e484a9f26d78799d090c0 : FAILURE in 8m 40s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/777e5a89d09e4636a6d350f850c0eedc : FAILURE in 11m 07s","accounts_in_message":[],"_revision_number":2},{"id":"436434b143f89f2a3c1ec79b1fcbfb28e40ece72","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2024-08-02 07:19:47.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f5effbd2436ffd3e0fecd4e7bcc5205ad99cc278","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2024-08-19 19:09:08.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f226c215b370dee082a5fbe9d36ca7360503a1f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-19 20:10:56.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0b6c8c0c9c8c4083a1f58c552556ec49\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/703d623e97e94e329efd105dc9d78b43 : SUCCESS in 3m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b79cf76adc3a4d56bfb6db63436d19ff : SUCCESS in 5m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/67b981b8b28846278c54d65c15018013 : SUCCESS in 4m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ac0284277e804f0789b598a1653de152 : SUCCESS in 5m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ad25ada0f6244de83bb9c7a2ba6b259 : SUCCESS in 6m 56s\n- requirements-check https://zuul.opendev.org/t/openstack/build/492bd6be5e2f4efab908853111e39268 : SUCCESS in 2m 36s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/a6fa8978e4ba45c0b9624964bac4b58b : SUCCESS in 39m 48s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/0c91d253fb3f45d99713632f9c6856c8 : SUCCESS in 39m 31s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/8456c385f2804059819d2e88f01e94ae : SUCCESS in 51m 38s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/51d11211312641ee95d5a9d944931453 : SUCCESS in 43m 22s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/adb17f7b62f2466a9469127eb5be5406 : SUCCESS in 34m 01s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/6e6215650c0348b1bb387a6864f2f45a : SUCCESS in 38m 14s","accounts_in_message":[],"_revision_number":2},{"id":"6517647a8908ca13265d7a91fa5dad63f2bd2306","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2024-08-26 10:22:37.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a838998306fb9cb2e16e464685eaa1440aa18535","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-08-26 16:53:43.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c80d8dc231c49a83820c23347a5a7474e8921315","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2024-08-27 05:44:16.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"e1807437d4e71aaf146cfb26fab371db2a478686","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2024-08-27 05:44:39.000000000","message":"Patch Set 2: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7ceb5b27aee29d1712a0a061cd13d4fac4837ad2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-27 05:45:11.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"177bb94528d8f42386215a35f4dc20083718c2a7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-27 06:38:25.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/66d138cc776d4197a5cbd95bcfe56b8b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/110b10eec84d4f25ac6e1a8eccc914e2 : SUCCESS in 3m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/62d9947b39654ca1b29cdc3a2ef86183 : SUCCESS in 4m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/44a249a466d94b88ab9cd826d4e50c29 : SUCCESS in 5m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d8404da312b4dcfa8a6be890f2bc269 : SUCCESS in 5m 56s\n- requirements-check https://zuul.opendev.org/t/openstack/build/9322a9e5b626463fb6aa4cb6a34d3ee6 : SUCCESS in 2m 20s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/8b63e02cd7e240b19acdf1670c1914b1 : SUCCESS in 35m 08s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/2ddddf6212cb4433b6493c917ef83d93 : SUCCESS in 34m 10s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/182d58eb50a24eda9ce23ed5622fe155 : SUCCESS in 38m 20s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/e79d3a8111c94048a9cf106635c4c6ab : SUCCESS in 32m 57s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/65f74f5066204826936047e7e3d7a5ec : SUCCESS in 47m 23s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/4ee8eb7102504b6da7f330b127c29fe3 : SUCCESS in 39m 35s","accounts_in_message":[],"_revision_number":2},{"id":"7e4f9e700ee3fdeac657a22002d8477776ebd4f7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-27 06:38:25.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"a2f0a0c42eb8926d41a88d72250a4330ebacadd6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-27 06:39:31.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/28a2366193b64b61a4f184410c81f4c6\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2c0a7eff5db42a0860dc2c6fd5c7c81 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ab00dac26f44d85a946e45231ee659efdc2c548b","revisions":{"afd9d022107d315eb7eaec0099bd0b733c4a87ea":{"kind":"REWORK","_number":1,"created":"2024-08-01 19:27:00.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/29/925529/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/29/925529/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/29/925529/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/29/925529/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/29/925529/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/29/925529/1"}}},"commit":{"parents":[{"commit":"f43e90311724ad574ff46d0165ae7f30395b9aaf","subject":"Merge \"reno: Update master for unmaintained/xena\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/f43e90311724ad574ff46d0165ae7f30395b9aaf"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-08-01 19:25:45.000000000","tz":-420},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-08-01 19:26:59.000000000","tz":-420},"subject":"Keep new RBAC disable by default","message":"Keep new RBAC disable by default\n\noslo.policy has enabled the new RBAC config options\nenforce_scope and enforce_new_defaults by default[1][2].\n\nUntil Aodh is ready with the new RBAC, we need to continue\ndisable them by overriding their default values.\n\nNOTE: there is no change in behaviour, aodh continue with the\nold RBAC as default.\n\nNeeded-By: https://review.opendev.org/c/openstack/requirements/+/925464\n\n[1] https://review.opendev.org/c/openstack/oslo.policy/+/924283\n[2] https://review.opendev.org/c/openstack/releases/+/925032\n\nChange-Id: Ifac31d1fc11b44fd6d3d18041f8bacee8287bccf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/afd9d022107d315eb7eaec0099bd0b733c4a87ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/afd9d022107d315eb7eaec0099bd0b733c4a87ea"}]},"branch":"refs/heads/master"},"ab00dac26f44d85a946e45231ee659efdc2c548b":{"kind":"REWORK","_number":2,"created":"2024-08-01 19:32:04.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/29/925529/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/29/925529/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/29/925529/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/29/925529/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/29/925529/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/29/925529/2"}}},"commit":{"parents":[{"commit":"f43e90311724ad574ff46d0165ae7f30395b9aaf","subject":"Merge \"reno: Update master for unmaintained/xena\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/f43e90311724ad574ff46d0165ae7f30395b9aaf"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-08-01 19:25:45.000000000","tz":-420},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-08-01 19:31:49.000000000","tz":-420},"subject":"Keep new RBAC disable by default","message":"Keep new RBAC disable by default\n\noslo.policy has enabled the new RBAC config options\nenforce_scope and enforce_new_defaults by default[1][2].\n\nUntil Aodh is ready with the new RBAC, we need to continue\ndisable them by overriding their default values.\n\nNOTE: there is no change in behaviour, aodh continue with the\nold RBAC as default.\n\nNeeded-By: https://review.opendev.org/c/openstack/requirements/+/925464\n\n[1] https://review.opendev.org/c/openstack/oslo.policy/+/924283\n[2] https://review.opendev.org/c/openstack/releases/+/925032\n\nChange-Id: Ifac31d1fc11b44fd6d3d18041f8bacee8287bccf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/ab00dac26f44d85a946e45231ee659efdc2c548b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/ab00dac26f44d85a946e45231ee659efdc2c548b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
