)]}'
{"id":"openstack%2Faodh~925775","triplet_id":"openstack%2Faodh~master~I3fceb9a244f248f1e62a1267408c82d1b35a4d37","project":"openstack/aodh","branch":"master","topic":"aodh_metrics","attention_set":{},"removed_from_attention_set":{"32240":{"account":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},"last_update":"2025-04-09 16:35:39.000000000","reason":"Change was submitted"},"5202":{"account":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"last_update":"2025-04-09 16:35:39.000000000","reason":"Change was submitted"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-04-09 15:40:00.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},"4264":{"account":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"last_update":"2025-04-09 10:03:04.000000000","reason":"\u003cGERRIT_ACCOUNT_4264\u003e replied on the change","reason_account":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"}},"34975":{"account":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"last_update":"2025-04-09 16:35:39.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3fceb9a244f248f1e62a1267408c82d1b35a4d37","subject":"Add evaluation results counters to /metrics","status":"MERGED","created":"2024-08-06 14:05:27.000000000","updated":"2025-04-09 16:36:50.000000000","submitted":"2025-04-09 16:35:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":88,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"925775-aodh_metrics","meta_rev_id":"3e7580b6f6528c48e6e5894e1659423cd6cc9b5f","_number":925775,"virtual_id_number":925775,"owner":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-04-09 16:35:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"all":[{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2025-04-09 10:03:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":2,"date":"2025-03-31 11:05:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":1,"date":"2025-04-09 15:40:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"}],"reviewers":{"CC":[{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"}],"REVIEWER":[{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-06 15:01:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-08-14 07:30:22.000000000","updated_by":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"reviewer":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},"state":"REVIEWER"},{"updated":"2024-08-14 07:30:22.000000000","updated_by":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"},{"updated":"2024-08-14 07:30:22.000000000","updated_by":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"reviewer":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"state":"REVIEWER"},{"updated":"2024-10-16 10:20:27.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"},{"updated":"2024-11-19 13:43:34.000000000","updated_by":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2025-01-13 15:49:42.000000000","updated_by":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"reviewer":{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},"state":"CC"},{"updated":"2025-03-31 11:05:19.000000000","updated_by":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"reviewer":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"state":"REVIEWER"}],"messages":[{"id":"b401376da40a8658c8ab23cc8c7e754b5e3d8d5d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2024-08-06 14:05:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b0cdbe6ab801910ae08e44f717b04571c64a2f5b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-06 15:01:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a388587aebc442c496fdc8c64736e9a4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da386c8e967143eb8eb0b36ee4e48f56 : SUCCESS in 4m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/75c7f237529b4713b96f0ef11dfff8d5 : SUCCESS in 6m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7258b0de87854dbc8b47eb4b9380d0f6 : SUCCESS in 6m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/363cd8452f554cd48f01724e38bfdef4 : SUCCESS in 10m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34c2a51a17774a30b7785820644fd722 : SUCCESS in 7m 27s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/447a7006a75041f6b080b34d13a9182f : SUCCESS in 30m 36s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/f95713619aef4e38ab14d77cc6f97833 : SUCCESS in 37m 55s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/695c0fc4f2b44304b3d024bca3e5ea92 : SUCCESS in 39m 17s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/18a2f0c9ab214eba99da490d8d6d2d8e : SUCCESS in 34m 14s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/6791c0dd04c64f3d9693c12cd9e7e3eb : SUCCESS in 39m 39s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/7b8538328aa249c1bfb9a82d49ee0e4c : SUCCESS in 55m 50s","accounts_in_message":[],"_revision_number":1},{"id":"d05313ffda7e596032796f028c6a093486256bb9","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2024-08-19 06:24:18.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8c5f547141360975cd933f5109bcce399872a709","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2024-08-26 12:32:28.000000000","message":"Patch Set 1: -Code-Review","accounts_in_message":[],"_revision_number":1},{"id":"0a8cadf6acf5a990d2118e3f222577f209a619bc","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2024-09-04 16:09:00.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a5b2dad663a084cdb95ef1b9bdf464f6af92b7dc","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2024-09-04 16:35:03.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f4464092f8acded2478c17013ea883db925582b3","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2024-09-04 16:35:13.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"380e6ecc4213394f30ea6b9801d0f8cbccfc6adf","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2024-09-05 13:43:13.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"fd9036ab542a4ecdeaae6414ce0926066a6a4e61","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-05 14:51:36.000000000","message":"Patch Set 2: Verified-1\n\n(6 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9347a21c5ebb4c56923f51a4728516c8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e17923685f83498d85397a6f372c9712 : FAILURE in 4m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/eb9aab1d22404d41bf43efa17e1ee8c4 : FAILURE in 6m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5f6db02d540143b8bf6d01f1b5c2fb94 : FAILURE in 4m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/98521fb9dc754880a7976ef8d6dd8859 : FAILURE in 5m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a66b2dd1a3674cd3b6f164b048b6d0d2 : SUCCESS in 9m 32s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/2af64f1c6d7c4972ab1c533963b03c94 : SUCCESS in 33m 05s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/7972e3c82d924a71ba5001022de604bb : SUCCESS in 36m 22s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/bc8c3980c43443bda2eca37761224b77 : SUCCESS in 42m 28s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/2dc2ed7c72034a289c8e6dc1fc3f1a6e : SUCCESS in 35m 49s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/ccd227afa09a4595afd29a9bd277e984 : SUCCESS in 46m 58s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/b20340f319164fcb89bda9161c830ccf : SUCCESS in 58m 53s","accounts_in_message":[],"_revision_number":2},{"id":"9c400f4c555786d7edb90c3f992f64d474295cf9","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2024-09-05 17:57:00.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"37de84b2ccda57893834a099d93f95e85cab2eb7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-05 18:52:26.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/37b2b9ac3a66417e900f09932d9ed951\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2fdfaec22e8d4008a5661877c98d6146 : SUCCESS in 4m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/17018d62ffd04bfaad8e5ebead0338ae : SUCCESS in 4m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8499266620b6417d8406d1b44e2e92f3 : SUCCESS in 4m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/22e69a4834a24319a07e1812c0b6f896 : FAILURE in 5m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8fc37385ea034778a42868b39364554e : SUCCESS in 6m 37s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/f7baf9d0bf7e4cc1a0d44379fc592716 : SUCCESS in 24m 23s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/e87c84a178b74b5581dc5921ba25ed9c : SUCCESS in 38m 32s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/cc0b4730f55f45cbb9307e2d218baa1a : SUCCESS in 42m 47s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/998e4a287ffb4fc18bba098a1f1106d1 : SUCCESS in 45m 13s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/f49df532b090403398fc9a1269ad0965 : SUCCESS in 54m 47s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/3b38ff36327b45d38ecbfdb8613df409 : SUCCESS in 37m 42s","accounts_in_message":[],"_revision_number":3},{"id":"4d56f70612293c6ed88be7396e7cb662de0b48b9","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2024-09-27 07:07:52.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":3},{"id":"ebff3607c5146aa30abc7984647fc95eba03e0f2","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2024-09-30 16:14:56.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"fcd01c537313cfda13e1803e3cafb8d074e9a0bc","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2024-10-02 14:10:31.000000000","message":"Patch Set 3: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":3},{"id":"1464dd204e6842079b2dc6000455e3d8a2d3a1ad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2024-10-02 19:42:04.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"bede010080878f4ecb5fa1790d0ce6bf405f44c3","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2024-10-02 19:42:37.000000000","message":"Patch Set 4:\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"26029a52213780399fb82a46e84f89a09247b1cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-02 20:43:16.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/50ce16bd1d6248348c3aeff7dd480db6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9a26f6d5f4b4b508a13dda0bc6ec18e : SUCCESS in 3m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/98628d77aa29421488e804d46b89a399 : SUCCESS in 4m 37s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/64faaf120f844516b55de8ef5c07e531 : SUCCESS in 8m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7288723102a34d88b6e4f60513d810df : SUCCESS in 6m 21s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/91337dcb51964eaca7d6047041f3630d : SUCCESS in 27m 32s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/27a566bcfa7642fb8fb09e9bd2ac5a9f : FAILURE in 21m 23s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/4109c3d7d58247dbb60fed8f95211e92 : SUCCESS in 34m 27s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/f96c7d572a5e4699949ea9851a77273c : SUCCESS in 30m 47s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/35ab255838194bf1b6273f8d203be408 : SUCCESS in 33m 53s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/52ace1b110f4444ab17b0e97d01a2009 : SUCCESS in 56m 04s","accounts_in_message":[],"_revision_number":4},{"id":"42434200a0e3cc38fc6d5d517402cf58b9ffb669","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2024-10-03 15:04:00.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"55b19179fde4d952ec76639e281781b912707d31","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2024-10-04 09:01:31.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"d8d1adc4563c6c0832d5a112967d530a48515954","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2024-10-15 14:21:20.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"be42d19fb82a09bcece4b705f7469aa94ce141ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-15 15:33:29.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/665dc4d0ed9347a08b1ce5efa712f851\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e90fe72fb5434ae4ba640f27e2c12721 : SUCCESS in 3m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f44f81a6a7dd4fcf8bf86b647fdb7588 : SUCCESS in 4m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8e129228e8e44d75a2655214e47db203 : SUCCESS in 5m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a1831d9654c40d6a50af95c16b0ed47 : SUCCESS in 6m 40s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/bb60ca9eb25048f2902ff560df798330 : SUCCESS in 27m 51s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/625d962e00664c8e8861de82ba0ed8f6 : FAILURE in 29m 18s (non-voting)\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/ac582b73dfd74fc6a40e08002559c7ab : SUCCESS in 44m 21s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/7ba9e86bbfaa414b8dfbf2e01cf184a2 : SUCCESS in 29m 05s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/15c80f2dc3854d5d9a11011d9d9f62d4 : SUCCESS in 59m 09s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/8b5e8f257f8f4d3d88aeb04985072f8c : SUCCESS in 1h 00m 00s","accounts_in_message":[],"_revision_number":4},{"id":"eff3bffb842ce55eb2f775bb8e496aba1fd8fc8c","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-16 10:20:27.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"6ce9895fb93889725024be9e8682db851c95bfc4","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-16 10:43:45.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"24e766b29b3f109355aaaf95743376fa6a6f78de","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2024-10-16 11:56:50.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"56da708b470d336c154b51f7e1ff7b47b5897a4a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2024-11-07 20:00:41.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"b8481ec35252490b55ce31e2de62e1e4e482eea5","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2024-11-07 20:05:16.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"13e3448712629c4119978fdd930ffbc1842c39ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-07 21:06:33.000000000","message":"Patch Set 5: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2d177d64541a43ba8ddfc5a0be5731f6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5450afb251604dab86e1c875715e59fc : FAILURE in 3m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f150c01e21bb454a94d3ad798ecd8398 : FAILURE in 4m 49s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d2fa9c5d4acf452694197e198b0037e4 : FAILURE in 5m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99d037219f704084ab93fbfc5e998c20 : SUCCESS in 7m 33s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/f051db394d4e4e89be87524bc1ca0e1e : SUCCESS in 39m 02s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/84bf801046134f5889bc4077dbb98a72 : SUCCESS in 36m 24s (non-voting)\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/9bae4aff07064b78900d10d5821e9c72 : SUCCESS in 46m 39s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/d3588415ffd2414992855934605150d9 : SUCCESS in 43m 40s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/d3173b7e23a7428fac4f287ad9be9c4b : SUCCESS in 54m 41s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/09414b19ecfe4ef6ad451b3b155bf034 : SUCCESS in 58m 10s","accounts_in_message":[],"_revision_number":5},{"id":"5dfcab177ed362e7b755f71dea2acc6718587ebc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2024-11-07 22:04:28.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"cbfb5acba00779f12e2f6bf7983bf9569a663028","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-07 23:03:03.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e95e9c8c4e9f4cf0a54efc4523da5454\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d0171ffc0844471aa218d154498a6a7 : SUCCESS in 3m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/93cf311d35c04e1aa97e8d55bfc1f523 : SUCCESS in 4m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8de9110cfa804452960460868dbbfd3e : SUCCESS in 5m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7f3d164440e4b318294eaecf8409d31 : SUCCESS in 7m 05s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/3cc6c4b13b8e4a0bae3cc59b7c173f2d : SUCCESS in 42m 33s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/b521e165e2264e4394ea52fcd647a281 : SUCCESS in 39m 37s (non-voting)\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/e6d7b32f37eb4feab5608f828478d6d3 : SUCCESS in 48m 41s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/1de8bc01c2cd4f44a993302e919cc492 : SUCCESS in 52m 07s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/f275964d712c4111ba3a130a349b195d : SUCCESS in 43m 48s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/5502e417d339409e99a0d4d860964ed2 : SUCCESS in 57m 41s","accounts_in_message":[],"_revision_number":6},{"id":"d79267048886fca6ac1276740a44bb0a19a4724a","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2024-11-08 07:52:07.000000000","message":"Patch Set 6: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"23e3ae8fb95b01dcf6d8eaf13c81742cf3ddd30b","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2024-11-19 13:43:34.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"965d312e2d1586c4b87d0bb165a7c3b44e155c0f","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-19 13:58:01.000000000","message":"Patch Set 6: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":6},{"id":"317b3c10173410f20f83fc42adc7e3affc154479","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-19 14:01:40.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"20c5f67245b5559296b4c616b59280c9a7edee65","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-19 14:14:40.000000000","message":"Patch Set 6:\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"a588ccafdc1189bc4f10f6d34d2ecd29e9b81c2e","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-19 14:15:44.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"eddfa2c0346fdabbd0f2b767126cbbd1f094c0d2","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-19 14:17:53.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"531c339f3f72ee3cfe354e6aeb71f919117bc5f0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-19 14:53:41.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"68d42e31edbfb49a24d729dfede15a54746e7a53","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-19 14:54:18.000000000","message":"Patch Set 6: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"0bdd8aae5e1ef3c32ac58d3f608c3e349e5201ec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-19 14:56:27.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"b07257c305392c0f4d869dfbc0d71108d3333769","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-19 15:00:39.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"96d7a3e0462b8e9bde5988995efb564fd95deab7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-19 16:06:20.000000000","message":"Patch Set 8: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/869aa624c2c54703949634e54e0ef129\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f9ca3cc25444095b07fd79456f89196 : FAILURE in 3m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d566172bf6d24e44a814b066d02bc9b6 : SUCCESS in 4m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/480d0f3589ff430bbd5795b04ae08fd4 : SUCCESS in 6m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08a9d89e89984bfca0e6207610ba65b3 : SUCCESS in 7m 21s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/915bd91b7eae438d8deeb9b355c72d5b : SUCCESS in 32m 20s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/078ef5405c114da68ec2a69092590b3b : SUCCESS in 31m 35s (non-voting)\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/4c65f260b21c4087b9d1bef3fad4b7ab : FAILURE in 38m 33s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/fe6df4ab1ab042949b0fe66010c8487e : SUCCESS in 44m 35s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/31c6da7d9c3a47bb878fb7b49af02cea : FAILURE in 1h 02m 56s","accounts_in_message":[],"_revision_number":8},{"id":"6feb2ba5042d634136321e20cda9cf3c9d655f30","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-19 17:06:49.000000000","message":"Patch Set 9: Published edit on patch set 8.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"9f1cb98101720fdf4e0e62c707772f708aff1ef5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-19 18:24:59.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/50526031a6dd429db7a694901800b2e6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf05c82e2c4148e98c841d45ce96df31 : SUCCESS in 4m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/58100606e8824b1aa7c65fcb1999c7bd : SUCCESS in 4m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4db98750053e4f1cbbf77074a5e83d87 : SUCCESS in 6m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf055c48443a4b43bcd1d23f37572d1a : SUCCESS in 6m 28s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/226ef6bb352e4402a3421b358af382d1 : SUCCESS in 33m 53s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/ecb15d72b93b4e5d9472bf0b6f8362dc : SUCCESS in 23m 16s (non-voting)\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/03b889d7e21a4dd8a5924f5301d9eb41 : SUCCESS in 41m 38s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/0e0db265a7d44d6d9b8d9d0f3f9e4249 : SUCCESS in 46m 40s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/aa09bb42cf1449a29fcb5d9f5748d420 : SUCCESS in 1h 03m 38s","accounts_in_message":[],"_revision_number":9},{"id":"db074cb5c6a2967402d5e020b03cb0ecd15b3431","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-20 06:02:37.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"43a4d200fe787e342702d7eca1d610dde4ef1e1c","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2024-11-20 13:59:15.000000000","message":"Patch Set 9:\n\n(3 comments)","accounts_in_message":[],"_revision_number":9},{"id":"b7c71cc0270bd653c5fecdb4c9b8fc64f131c2e1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2024-11-20 14:35:02.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":10},{"id":"430e4ed87f89f57bf7ea84f81487fbb2eae895bf","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-20 14:46:27.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"6a64791ad2025f0deb62a794d07e4d862f5935cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-20 15:35:05.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/95043c5c307d4840a8f3fa648b78aa8d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/745da868492741f59293ec378f632075 : SUCCESS in 3m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ecaae7de30564b27864e52196ea2c3a5 : SUCCESS in 4m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/597bf812a3e949969b761f261d157389 : SUCCESS in 5m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50b7516112f344de97beda2e639886ff : SUCCESS in 6m 59s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/d9e23ba49a8c45d58bdc8585babdabdc : SUCCESS in 40m 18s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/3e7b4dd35d614d75869595cca962676d : SUCCESS in 32m 32s (non-voting)\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/c661648bb0da4f9c8d272e76cf2b297a : SUCCESS in 38m 03s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/53176ea6c2d14e548e18068c1367cfa8 : SUCCESS in 54m 29s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/1acd36c2980845ab9d56f5dc3e5a6cbc : SUCCESS in 45m 32s","accounts_in_message":[],"_revision_number":10},{"id":"b7ad318153389c4e8d09c5fe763b9ee7cd35ad77","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-20 15:56:56.000000000","message":"Patch Set 10: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"07332194bfa1c03d7c5efb3a606ffcc756e5d64e","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-20 15:57:48.000000000","message":"Patch Set 10: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"2fe4ae483cf2fae0e6112a421bf80d4daddbeeab","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2024-11-20 21:28:59.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"463c24df34c9f5b89867827a0f89bb63168677e1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2024-11-20 21:40:02.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":11},{"id":"5d9e02a9e3c0860a027d49c10e6353f8062b731e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-20 22:37:56.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7c78378b1b2049ebbd85e39baaa9c782\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/81efe0f77986494eadb218f055a15854 : SUCCESS in 5m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4ba082e43a7848f282cbf89aaf3850fc : SUCCESS in 4m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/815397332b7640d98f866b79c3962633 : SUCCESS in 3m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d0f45c2135f4050b80ca0fec9374e13 : SUCCESS in 6m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/657ca721735b439a8e5fdc5e1b21dafe : SUCCESS in 7m 28s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/87badd5cc2e045c5b65467cd1711a24c : SUCCESS in 29m 34s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/7faa09e1995f47b593dd61d69223e562 : SUCCESS in 22m 29s (non-voting)\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/b1f9f59ca30b41acb4e0202b0303b116 : SUCCESS in 30m 39s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/1415b08354b344d4b126d7867d22197f : SUCCESS in 40m 59s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/070359b2bc1b41b58203051788dcadf1 : SUCCESS in 50m 29s","accounts_in_message":[],"_revision_number":11},{"id":"43a96ee92168c68d8dfe92e275a3d5f219beaf2d","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-22 11:44:47.000000000","message":"Patch Set 11: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":11},{"id":"f2cae12a196471ad3fa5e39c53211e506b60739b","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2025-01-13 15:47:08.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":11},{"id":"b6bd0fa2c38eeb2bcee1f8bd2d6a7f7fc165f9d0","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2025-01-13 15:49:42.000000000","message":"Patch Set 11: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"962d256f8aeed5754636f000057b0e296c0f182e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2025-01-16 18:35:49.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":12},{"id":"29ff594594951727e66c85cce6de5f7110476ac2","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2025-01-16 18:40:07.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":12},{"id":"087f4858a9dc41b061ddfbe66e59d498bc0caa90","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2025-01-16 18:42:17.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"934f30a2b3990559db53c05bd1c822aadd626034","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2025-01-16 18:45:25.000000000","message":"Topic add_counters removed","accounts_in_message":[],"_revision_number":12},{"id":"59e13c9b02cf259d6d6b85307f7bfac2f319c31d","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2025-01-16 18:45:37.000000000","message":"Topic set to aodh_metrics","accounts_in_message":[],"_revision_number":12},{"id":"b69e94f78b3f05bf0c271839db9f475d391f5ead","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2025-01-16 19:20:41.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"866af6e2cfc86bf8e93e4490af5d75c416a69596","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-16 20:20:45.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2ab8f993ccb549e48c1052e1ee21652f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/961bde9a06ea4f4ca6ea64c87a953ab2 : SUCCESS in 4m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2e15ed83a4c54c8383e2d8ced43352a1 : SUCCESS in 4m 05s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9e81877c4a9f47e1a32ce43b24502518 : SUCCESS in 5m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5fb34c8971af4ffc877dbd9802fb54c8 : SUCCESS in 6m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/81e5f1f0e0264bc6b03c543a9c32ac36 : SUCCESS in 5m 19s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/2c5f343d07004eb69cd07ec6dc97bd2b : SUCCESS in 40m 20s\n- aodh-tempest-plugin-mysql-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3e4bf6c33a114469a14a0324809a3825 : SUCCESS in 24m 55s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/1422bb07961f4814923ff45fdb7bd9c9 : SUCCESS in 26m 53s (non-voting)\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/5300b81f083c4a0fb015589f88f17e7e : SUCCESS in 35m 49s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/ec8caeb06b874fd4af46ca0330133ed6 : SUCCESS in 54m 27s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/5e2400ebaa464256af231975f643fdae : SUCCESS in 53m 12s","accounts_in_message":[],"_revision_number":13},{"id":"3cf68da35dda335bc45de712775c70da6471ef17","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2025-01-21 19:59:34.000000000","message":"Topic add_counters removed","accounts_in_message":[],"_revision_number":13},{"id":"9cf8587a695a449a2e8e3ce909ec00f51284c2ca","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2025-01-21 19:59:42.000000000","message":"Topic set to aodh_metrics","accounts_in_message":[],"_revision_number":13},{"id":"112f430bc1e1283c72d9ea900e160d17c109c11d","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2025-01-23 09:32:43.000000000","message":"Patch Set 13: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":13},{"id":"a83b8daf0b6abe1c431ed587064c450df84af3b7","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2025-01-23 10:30:31.000000000","message":"Patch Set 13:\n\n(2 comments)","accounts_in_message":[],"_revision_number":13},{"id":"b89738597caca19361dcb1e7ed6e94d61653e95d","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2025-01-23 11:18:06.000000000","message":"Patch Set 13:\n\n(2 comments)","accounts_in_message":[],"_revision_number":13},{"id":"cd0ec930c041da0f28e5190eefa0020f9e0b7fe1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2025-01-29 08:51:14.000000000","message":"Uploaded patch set 14.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":14},{"id":"68c3cec026e263d11c845d704cef2170fef3f210","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2025-01-29 08:52:07.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"a0e65a276d6862d034d9f19d2e94b778fb82e750","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-29 10:14:20.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/66ac18cee5bf48b080819d877294f74c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c5a498b5a9345a896586e2f0f92cd91 : SUCCESS in 4m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5d62b3547b014d93b75c41f63eb48677 : SUCCESS in 4m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7175ea03de754ab38d2a9feaa571a8a9 : SUCCESS in 5m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1fc3a8ed21e4735a6fda950cf8208d5 : TIMED_OUT in 30m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bdcdcf798a664a98a1de1cf1a22c5825 : SUCCESS in 5m 56s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/1ce37069a53f43a88488c0227da7a01f : SUCCESS in 28m 46s\n- aodh-tempest-plugin-mysql-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4e75e87ae6994f1a9cc7cee205fbb4f3 : SUCCESS in 39m 14s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/de60cf2866d94ea382d360a57f28de25 : SUCCESS in 39m 12s (non-voting)\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/4284de969a96407cb6c8bf3ea85e04e7 : SUCCESS in 48m 01s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/296fdff1e77b42b0be8915b5c3b38048 : FAILURE in 59m 42s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/cea6524cd8d94e5d8915dfd982764968 : FAILURE in 1h 03m 53s","accounts_in_message":[],"_revision_number":14},{"id":"7a184e084e91f53860b9eccaaca613e0d32be36e","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2025-02-05 10:40:24.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"a8f2f837e36a9dbd6350581c9ee83ffd9a1982c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-05 11:32:30.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/35a2ee82b95b4645b86a110cf64ca486\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f6a83562726242dbbcd0768b3433184f : SUCCESS in 4m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9d5e6496342c44e6a81fafb22dc61c10 : SUCCESS in 4m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2a04896ea7cb4b0aa19d0bfbb4e04cc9 : SUCCESS in 5m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f7bd4171f8f4670978c61aed374e4e7 : SUCCESS in 6m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/55b55b04bb8d4439a6f0d44e61105c14 : SUCCESS in 7m 08s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/e5ac3aacb1414032a32da735df41f397 : SUCCESS in 36m 08s\n- aodh-tempest-plugin-mysql-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/495fd3128eb44a68a945dfce5e02a27e : SUCCESS in 43m 52s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/3a0efe396d7f48cb94ef7d7594d2e6ef : SUCCESS in 28m 40s (non-voting)\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/4a81f34c07e44c11b20ab240a79ea7be : SUCCESS in 36m 44s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/c45567f521ba41ba83eee41d935dd424 : SUCCESS in 41m 37s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/241a937934fc476d948289bf026a01a4 : SUCCESS in 34m 53s","accounts_in_message":[],"_revision_number":14},{"id":"48d18f7509dd9730204a9dd197a89696c754ba82","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2025-02-06 08:17:10.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"29c9dc406f10464d60fd60803dea364b7d617b76","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2025-02-06 08:38:24.000000000","message":"Patch Set 14: Code-Review+2","accounts_in_message":[],"_revision_number":14},{"id":"5319eba249380511f39ca3af21331ede9c7c9b56","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2025-02-20 15:05:49.000000000","message":"Uploaded patch set 15.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":15},{"id":"28013ba6518ad596e5ee10810d3589a3b7fdf0d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-20 16:04:50.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f807030cecb04fcc86017aa03763d071\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/daa5c4da423e472bbd24287afb9fd968 : SUCCESS in 4m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e4b25d84642f4ca98a4295c38e852e77 : SUCCESS in 7m 06s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e93d26ffdce64b2b883bea0ae4d57ebd : SUCCESS in 5m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98d35327ef3d425e8651a2cf6c244fdd : SUCCESS in 8m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/71c4c5d039d14ff48f689887af1612f9 : SUCCESS in 6m 08s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/4457046196394847bd58dd264c7b4f9f : SUCCESS in 45m 44s\n- aodh-tempest-plugin-mysql-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/396f475607ac4657a71d079520576b56 : SUCCESS in 36m 44s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/2753099d5a0346bca8a768a423ea7fe8 : SUCCESS in 34m 52s (non-voting)\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/c48efc0f16fc4e43a6de9f8b6f2a3aa5 : SUCCESS in 37m 10s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/c19a0797f64948c9873d01431c68fef0 : SUCCESS in 50m 21s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/156a1d86d0984ae393fccacea7b51c64 : SUCCESS in 52m 05s","accounts_in_message":[],"_revision_number":15},{"id":"19db027942878590e96c746f8dc30bba2dad3890","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2025-02-20 16:19:19.000000000","message":"Patch Set 15: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"d883844eac6b640f41ab34ca13597434937710e2","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2025-02-24 13:29:03.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"076f86a69a2e2b7a5b5a28a43ba73c8d79c0a567","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-02-25 10:25:05.000000000","message":"Patch Set 15:\n\n(3 comments)","accounts_in_message":[],"_revision_number":15},{"id":"03d57e0b8b7b6f49573d13b5822660cc99abba42","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-02-25 10:41:08.000000000","message":"Patch Set 15: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":15},{"id":"8cf1cc8578a8d40d85a525fd5eaa9c1112d741e9","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2025-03-03 10:11:08.000000000","message":"Patch Set 15:\n\n(2 comments)","accounts_in_message":[],"_revision_number":15},{"id":"8d1c18cb31799e1742623345bdd842480da8c140","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-03-03 10:21:34.000000000","message":"Patch Set 15:\n\n(2 comments)","accounts_in_message":[],"_revision_number":15},{"id":"9233540c9fa8aa0d83cffaaee134f8904ba18c3e","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-03-03 10:32:18.000000000","message":"Patch Set 15:\n\n(2 comments)","accounts_in_message":[],"_revision_number":15},{"id":"15a2f282866b718da9f2e43481f364dfb644e4ee","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-03-03 10:45:49.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"dffad7b1dce030c4a6bdbc9355089676561c9a06","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-03-03 10:53:52.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"339533986941e6016b617879af4c6ac30c6b1149","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2025-03-04 15:10:41.000000000","message":"Uploaded patch set 16.\n\nOutdated Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":16},{"id":"028ad8cb11f78887846d655d9d4ffa1c09619824","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2025-03-04 15:14:58.000000000","message":"Patch Set 16:\n\n(7 comments)","accounts_in_message":[],"_revision_number":16},{"id":"54d288223fbb2214799e5110051bcf8f27aa06dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-04 16:56:00.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b4ed08ab0e194874be726e62a4fe3e34\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1bd1aaa7deea4f8e9844eaf7359e8f2c : SUCCESS in 6m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3b771bfd3ed346eaa1686c47a7fdfa7f : SUCCESS in 6m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/21b4944ec699419ab8b89a7c3673f6ac : SUCCESS in 6m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df87fa4581d04ad79c040b7d2b4a0926 : SUCCESS in 8m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3931bb4692c341f89c86196348a6912a : SUCCESS in 9m 02s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/e3f83b427db748f4a360a5731b986099 : SUCCESS in 37m 20s\n- aodh-tempest-plugin-mysql-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e8b81d93140743e58b10b60f4a5cfc64 : SUCCESS in 37m 11s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/e0222468476c4e68b2a1a1c09e1877b2 : SUCCESS in 1h 08m 32s (non-voting)\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/334a1bcfdd8a427cbda2a0ef28d04741 : SUCCESS in 48m 56s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/0b1308d8c8734842b93200d2e52da89f : SUCCESS in 1h 39m 24s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/e03f251de19a4cb7afb76a5c435de396 : SUCCESS in 59m 55s","accounts_in_message":[],"_revision_number":16},{"id":"e1ecafa78b9cf7871476ea1ae20fe7e21782745f","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-03-06 14:37:09.000000000","message":"Patch Set 16:\n\n(2 comments)","accounts_in_message":[],"_revision_number":16},{"id":"283effcd2e0b6ae891647bc946e9b5d2ed68a2a1","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-03-06 14:37:24.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"4b10f84334a93549a779f8f8629cf31aa1d8bb2b","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-03-06 14:38:39.000000000","message":"Patch Set 16:\n\n(2 comments)","accounts_in_message":[],"_revision_number":16},{"id":"1161494aab49ae5955e97d61c604a63791b67a3f","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2025-03-06 17:53:56.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"5cfa4c940a94d4e2ea241c24f5bec00d43b66498","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2025-03-18 08:57:08.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"a651404feec135f83fffbf99e6f4d370f599c318","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-03-18 09:05:59.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"52bb46635af33ab7ef66e33239c20a0147390689","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2025-03-18 09:08:51.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"4289f47275eaab3cc59eb7c3d2051c08d95dfd9b","author":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"date":"2025-03-31 11:05:19.000000000","message":"Patch Set 16: Code-Review+2","accounts_in_message":[],"_revision_number":16},{"id":"62c1db79c3eabe510931d01833881a4b51ecedcd","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2025-04-09 09:04:32.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"b992ac019729cf57123b1e95df31ca36f171460d","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2025-04-09 10:03:04.000000000","message":"Patch Set 16: Code-Review+2","accounts_in_message":[],"_revision_number":16},{"id":"d7569231087f6685cbb18946299bde1b3bdbde2d","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-04-09 15:40:00.000000000","message":"Patch Set 16: Workflow+1","accounts_in_message":[],"_revision_number":16},{"id":"9ff476094bec7a4843a4dcc8cc7c8111ccedf1bf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-09 15:40:45.000000000","message":"Patch Set 16: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":16},{"id":"6ac160745cbc3f24f3091682b2838e4694b5f4de","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-09 16:35:38.000000000","message":"Patch Set 16: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/846ea2c0f41b42f68181ed3734e21592\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b40cf8a7a9b47c99358493efdd67f4f : SUCCESS in 4m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5fe99d7a3cf64425b4da3b2f55eca2eb : SUCCESS in 4m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/388e6e318f17448090258697a1d0a818 : SUCCESS in 5m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f3c727774e3452baecbef51064aeb19 : SUCCESS in 8m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/96d4d82e289a411fb2d914da3a82807d : SUCCESS in 4m 51s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/d206a21854ff4811b57bb74894b2e9bd : SUCCESS in 37m 11s\n- aodh-tempest-plugin-mysql-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ccc69b53b246467fa5620fb725bdb765 : SUCCESS in 33m 12s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/bb8e89d3ef644418b0dcd3897f7a29fa : SUCCESS in 32m 23s (non-voting)\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/6d9f41fe313e436bbf54bf19bcc492f4 : SUCCESS in 41m 22s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/d775ce70a3b24a618f7705144b241870 : SUCCESS in 48m 27s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/3dac12a16672434e803a92f9a2071a44 : SUCCESS in 50m 08s","accounts_in_message":[],"_revision_number":16},{"id":"41d6599a75a07b48dc6f2d96f926eac21f0da58c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-09 16:35:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":16},{"id":"3e7580b6f6528c48e6e5894e1659423cd6cc9b5f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-09 16:36:50.000000000","message":"Patch Set 16:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9c2c15b156de44958db550516909e17c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7858caa51c444338acf1b2f02b5d64be : SUCCESS in 45s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/420d6251525d4b9fba9a16d58b91536a : SUCCESS in 52s","accounts_in_message":[],"_revision_number":16}],"current_revision_number":16,"current_revision":"1643abed422374733b3e73b4dcc701fd51114b84","revisions":{"435700d9dc128fb4085594eb1ab69a087b416341":{"kind":"REWORK","_number":1,"created":"2024-08-06 14:05:27.000000000","uploader":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"ref":"refs/changes/75/925775/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/75/925775/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/75/925775/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/75/925775/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/75/925775/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/75/925775/1"}}},"commit":{"parents":[{"commit":"f43e90311724ad574ff46d0165ae7f30395b9aaf","subject":"Merge \"reno: Update master for unmaintained/xena\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/f43e90311724ad574ff46d0165ae7f30395b9aaf"}]}],"author":{"name":"Jaromir Wysoglad","email":"jwysogla@redhat.com","date":"2024-08-06 14:01:56.000000000","tz":-240},"committer":{"name":"Jaromir Wysoglad","email":"jwysogla@redhat.com","date":"2024-08-06 14:02:33.000000000","tz":-240},"subject":"Add evaluation results counters to /metrics","message":"Add evaluation results counters to /metrics\n\nThis change adds code for collection and storage of\nevaluation result counters. The values of the counters can\nthen be retreived through a new /metrics api endpoint.\n\nChange-Id: I3fceb9a244f248f1e62a1267408c82d1b35a4d37\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/435700d9dc128fb4085594eb1ab69a087b416341"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/435700d9dc128fb4085594eb1ab69a087b416341"}]},"branch":"refs/heads/master"},"f21e11ad4809ced8ff0d8372fec1c945d85f7296":{"kind":"REWORK","_number":2,"created":"2024-09-05 13:43:13.000000000","uploader":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"ref":"refs/changes/75/925775/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/75/925775/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/75/925775/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/75/925775/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/75/925775/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/75/925775/2"}}},"commit":{"parents":[{"commit":"f43e90311724ad574ff46d0165ae7f30395b9aaf","subject":"Merge \"reno: Update master for unmaintained/xena\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/f43e90311724ad574ff46d0165ae7f30395b9aaf"}]}],"author":{"name":"Jaromir Wysoglad","email":"jwysogla@redhat.com","date":"2024-08-06 14:01:56.000000000","tz":-240},"committer":{"name":"Jaromir Wysoglad","email":"jwysogla@redhat.com","date":"2024-09-05 00:48:57.000000000","tz":-240},"subject":"Add evaluation results counters to /metrics","message":"Add evaluation results counters to /metrics\n\nThis change adds code for collection and storage of\nevaluation result counters. The values of the counters can\nthen be retreived through a new /metrics api endpoint.\n\nCounters are collected and served for each project separately.\nTo enable this feature set \"enable_metrics\" to true inside [DEFAULT].\n\nChange-Id: I3fceb9a244f248f1e62a1267408c82d1b35a4d37\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/f21e11ad4809ced8ff0d8372fec1c945d85f7296"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/f21e11ad4809ced8ff0d8372fec1c945d85f7296"}]},"branch":"refs/heads/master"},"c6ad61152cadd57e95117d0ca5e9c6c8eda3c85c":{"kind":"REWORK","_number":3,"created":"2024-09-05 17:57:00.000000000","uploader":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"ref":"refs/changes/75/925775/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/75/925775/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/75/925775/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/75/925775/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/75/925775/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/75/925775/3"}}},"commit":{"parents":[{"commit":"f43e90311724ad574ff46d0165ae7f30395b9aaf","subject":"Merge \"reno: Update master for unmaintained/xena\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/f43e90311724ad574ff46d0165ae7f30395b9aaf"}]}],"author":{"name":"Jaromir Wysoglad","email":"jwysogla@redhat.com","date":"2024-08-06 14:01:56.000000000","tz":-240},"committer":{"name":"Jaromir Wysoglad","email":"jwysogla@redhat.com","date":"2024-09-05 05:02:38.000000000","tz":-240},"subject":"Add evaluation results counters to /metrics","message":"Add evaluation results counters to /metrics\n\nThis change adds code for collection and storage of\nevaluation result counters. The values of the counters can\nthen be retreived through a new /metrics api endpoint.\n\nCounters are collected and served for each project separately.\nTo enable this feature set \"enable_metrics\" to true inside [DEFAULT].\n\nChange-Id: I3fceb9a244f248f1e62a1267408c82d1b35a4d37\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/c6ad61152cadd57e95117d0ca5e9c6c8eda3c85c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/c6ad61152cadd57e95117d0ca5e9c6c8eda3c85c"}]},"branch":"refs/heads/master"},"d69b1c0b330295517003408b4a01d722d8666da3":{"kind":"REWORK","_number":4,"created":"2024-10-02 19:42:04.000000000","uploader":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"ref":"refs/changes/75/925775/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/75/925775/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/75/925775/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/75/925775/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/75/925775/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/75/925775/4"}}},"commit":{"parents":[{"commit":"6f7ba5bbdaff544260966208f1ab54c75deffcca","subject":"Merge \"Simplify tox targets\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/6f7ba5bbdaff544260966208f1ab54c75deffcca"}]}],"author":{"name":"Jaromir Wysoglad","email":"jwysogla@redhat.com","date":"2024-08-06 14:01:56.000000000","tz":-240},"committer":{"name":"Jaromir Wysoglad","email":"jwysogla@redhat.com","date":"2024-10-02 19:41:37.000000000","tz":-240},"subject":"Add evaluation results counters to /metrics","message":"Add evaluation results counters to /metrics\n\nThis change adds code for collection and storage of\nevaluation result counters. The values of the counters can\nthen be retreived through a new /metrics api endpoint.\n\nCounters are collected and served for each project separately.\nTo enable this feature set \"enable_metrics\" to true inside [DEFAULT].\n\nChange-Id: I3fceb9a244f248f1e62a1267408c82d1b35a4d37\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/d69b1c0b330295517003408b4a01d722d8666da3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/d69b1c0b330295517003408b4a01d722d8666da3"}]},"branch":"refs/heads/master"},"eb71ed7d6da01fb80d468be9207c26a76ae3decd":{"kind":"REWORK","_number":5,"created":"2024-11-07 20:00:41.000000000","uploader":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"ref":"refs/changes/75/925775/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/75/925775/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/75/925775/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/75/925775/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/75/925775/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/75/925775/5"}}},"commit":{"parents":[{"commit":"6f7ba5bbdaff544260966208f1ab54c75deffcca","subject":"Merge \"Simplify tox targets\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/6f7ba5bbdaff544260966208f1ab54c75deffcca"}]}],"author":{"name":"Jaromir Wysoglad","email":"jwysogla@redhat.com","date":"2024-08-06 14:01:56.000000000","tz":-240},"committer":{"name":"Jaromir Wysoglad","email":"jwysogla@redhat.com","date":"2024-11-07 19:58:54.000000000","tz":-300},"subject":"Add evaluation results counters to /metrics","message":"Add evaluation results counters to /metrics\n\nThis change adds code for collection and storage of\nevaluation result counters. The values of the counters can\nthen be retreived through a new /metrics api endpoint.\n\nCounters are collected and served for each project separately.\nTo enable this feature set \"enable_metrics\" to true inside [DEFAULT].\n\nChange-Id: I3fceb9a244f248f1e62a1267408c82d1b35a4d37\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/eb71ed7d6da01fb80d468be9207c26a76ae3decd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/eb71ed7d6da01fb80d468be9207c26a76ae3decd"}]},"branch":"refs/heads/master"},"974867877b5de189a123b6712780a19a92ab058d":{"kind":"REWORK","_number":6,"created":"2024-11-07 22:04:28.000000000","uploader":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"ref":"refs/changes/75/925775/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/75/925775/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/75/925775/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/75/925775/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/75/925775/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/75/925775/6"}}},"commit":{"parents":[{"commit":"6f7ba5bbdaff544260966208f1ab54c75deffcca","subject":"Merge \"Simplify tox targets\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/6f7ba5bbdaff544260966208f1ab54c75deffcca"}]}],"author":{"name":"Jaromir Wysoglad","email":"jwysogla@redhat.com","date":"2024-08-06 14:01:56.000000000","tz":-240},"committer":{"name":"Jaromir Wysoglad","email":"jwysogla@redhat.com","date":"2024-11-07 22:02:53.000000000","tz":-300},"subject":"Add evaluation results counters to /metrics","message":"Add evaluation results counters to /metrics\n\nThis change adds code for collection and storage of\nevaluation result counters. The values of the counters can\nthen be retreived through a new /metrics api endpoint.\n\nCounters are collected and served for each project separately.\nTo enable this feature set \"enable_metrics\" to true inside [DEFAULT].\n\nChange-Id: I3fceb9a244f248f1e62a1267408c82d1b35a4d37\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/974867877b5de189a123b6712780a19a92ab058d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/974867877b5de189a123b6712780a19a92ab058d"}]},"branch":"refs/heads/master"},"09c92726b0b69cd3d4f37610117349d5e44404fd":{"kind":"REWORK","_number":7,"created":"2024-11-19 14:53:41.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/75/925775/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/75/925775/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/75/925775/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/75/925775/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/75/925775/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/75/925775/7"}}},"commit":{"parents":[{"commit":"6f7ba5bbdaff544260966208f1ab54c75deffcca","subject":"Merge \"Simplify tox targets\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/6f7ba5bbdaff544260966208f1ab54c75deffcca"}]}],"author":{"name":"Jaromir Wysoglad","email":"jwysogla@redhat.com","date":"2024-08-06 14:01:56.000000000","tz":-240},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-19 14:49:18.000000000","tz":540},"subject":"Add evaluation results counters to /metrics","message":"Add evaluation results counters to /metrics\n\nThis change adds code for collection and storage of\nevaluation result counters. The values of the counters can\nthen be retreived through a new /metrics api endpoint.\n\nCounters are collected and served for each project separately.\nTo enable this feature set \"enable_metrics\" to true inside [DEFAULT].\n\nChange-Id: I3fceb9a244f248f1e62a1267408c82d1b35a4d37\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/09c92726b0b69cd3d4f37610117349d5e44404fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/09c92726b0b69cd3d4f37610117349d5e44404fd"}]},"branch":"refs/heads/master"},"19035762ea2a2dfbc9eba4c644c32fa72889901a":{"kind":"REWORK","_number":8,"created":"2024-11-19 14:56:27.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/75/925775/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/75/925775/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/75/925775/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/75/925775/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/75/925775/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/75/925775/8"}}},"commit":{"parents":[{"commit":"6f7ba5bbdaff544260966208f1ab54c75deffcca","subject":"Merge \"Simplify tox targets\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/6f7ba5bbdaff544260966208f1ab54c75deffcca"}]}],"author":{"name":"Jaromir Wysoglad","email":"jwysogla@redhat.com","date":"2024-08-06 14:01:56.000000000","tz":-240},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-19 14:56:18.000000000","tz":540},"subject":"Add evaluation results counters to /metrics","message":"Add evaluation results counters to /metrics\n\nThis change adds code for collection and storage of\nevaluation result counters. The values of the counters can\nthen be retreived through a new /metrics api endpoint.\n\nCounters are collected and served for each project separately.\nTo enable this feature set \"enable_metrics\" to true inside [DEFAULT].\n\nChange-Id: I3fceb9a244f248f1e62a1267408c82d1b35a4d37\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/19035762ea2a2dfbc9eba4c644c32fa72889901a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/19035762ea2a2dfbc9eba4c644c32fa72889901a"}]},"branch":"refs/heads/master"},"a1f31c6750120f40da92bf9335511fd068a63289":{"kind":"REWORK","_number":9,"created":"2024-11-19 17:06:49.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/75/925775/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/75/925775/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/75/925775/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/75/925775/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/75/925775/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/75/925775/9"}}},"commit":{"parents":[{"commit":"6f7ba5bbdaff544260966208f1ab54c75deffcca","subject":"Merge \"Simplify tox targets\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/6f7ba5bbdaff544260966208f1ab54c75deffcca"}]}],"author":{"name":"Jaromir Wysoglad","email":"jwysogla@redhat.com","date":"2024-08-06 14:01:56.000000000","tz":-240},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-19 17:06:49.000000000","tz":0},"subject":"Add evaluation results counters to /metrics","message":"Add evaluation results counters to /metrics\n\nThis change adds code for collection and storage of\nevaluation result counters. The values of the counters can\nthen be retreived through a new /metrics api endpoint.\n\nCounters are collected and served for each project separately.\nTo enable this feature set \"enable_metrics\" to true inside [DEFAULT].\n\nChange-Id: I3fceb9a244f248f1e62a1267408c82d1b35a4d37\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/a1f31c6750120f40da92bf9335511fd068a63289"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/a1f31c6750120f40da92bf9335511fd068a63289"}]},"branch":"refs/heads/master"},"13f52aa2f77fb66107eae346fdd782c9c2131a83":{"kind":"REWORK","_number":10,"created":"2024-11-20 14:35:02.000000000","uploader":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"ref":"refs/changes/75/925775/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/75/925775/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/75/925775/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/75/925775/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/75/925775/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/75/925775/10"}}},"commit":{"parents":[{"commit":"6f7ba5bbdaff544260966208f1ab54c75deffcca","subject":"Merge \"Simplify tox targets\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/6f7ba5bbdaff544260966208f1ab54c75deffcca"}]}],"author":{"name":"Jaromir Wysoglad","email":"jwysogla@redhat.com","date":"2024-08-06 14:01:56.000000000","tz":-240},"committer":{"name":"Jaromir Wysoglad","email":"jwysogla@redhat.com","date":"2024-11-20 14:32:23.000000000","tz":-300},"subject":"Add evaluation results counters to /metrics","message":"Add evaluation results counters to /metrics\n\nThis change adds code for collection and storage of\nevaluation result counters. The values of the counters can\nthen be retreived through a new /metrics api endpoint.\n\nCounters are collected and served for each project separately.\nTo enable this feature set \"enable_metrics\" to true inside [DEFAULT].\n\nChange-Id: I3fceb9a244f248f1e62a1267408c82d1b35a4d37\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/13f52aa2f77fb66107eae346fdd782c9c2131a83"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/13f52aa2f77fb66107eae346fdd782c9c2131a83"}]},"branch":"refs/heads/master"},"52fa96700d47724eef1c3cb80598657000e5a7ad":{"kind":"REWORK","_number":11,"created":"2024-11-20 21:40:02.000000000","uploader":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"ref":"refs/changes/75/925775/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/75/925775/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/75/925775/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/75/925775/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/75/925775/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/75/925775/11"}}},"commit":{"parents":[{"commit":"6f7ba5bbdaff544260966208f1ab54c75deffcca","subject":"Merge \"Simplify tox targets\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/6f7ba5bbdaff544260966208f1ab54c75deffcca"}]}],"author":{"name":"Jaromir Wysoglad","email":"jwysogla@redhat.com","date":"2024-08-06 14:01:56.000000000","tz":-240},"committer":{"name":"Jaromir Wysoglad","email":"jwysogla@redhat.com","date":"2024-11-20 21:37:42.000000000","tz":-300},"subject":"Add evaluation results counters to /metrics","message":"Add evaluation results counters to /metrics\n\nThis change adds code for collection and storage of\nevaluation result counters. The values of the counters can\nthen be retreived through a new /metrics api endpoint.\n\nCounters are collected and served for each project separately.\nTo enable this feature set \"enable_metrics\" to true inside [DEFAULT].\n\nChange-Id: I3fceb9a244f248f1e62a1267408c82d1b35a4d37\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/52fa96700d47724eef1c3cb80598657000e5a7ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/52fa96700d47724eef1c3cb80598657000e5a7ad"}]},"branch":"refs/heads/master"},"b3fbe13b305f354e7b3b6b2ecd091fefcbb13ede":{"kind":"REWORK","_number":12,"created":"2025-01-16 18:35:49.000000000","uploader":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"ref":"refs/changes/75/925775/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/75/925775/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/75/925775/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/75/925775/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/75/925775/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/75/925775/12"}}},"commit":{"parents":[{"commit":"6f7ba5bbdaff544260966208f1ab54c75deffcca","subject":"Merge \"Simplify tox targets\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/6f7ba5bbdaff544260966208f1ab54c75deffcca"}]}],"author":{"name":"Jaromir Wysoglad","email":"jwysogla@redhat.com","date":"2024-08-06 14:01:56.000000000","tz":-240},"committer":{"name":"Jaromir Wysoglad","email":"jwysogla@redhat.com","date":"2025-01-16 18:30:08.000000000","tz":-300},"subject":"Add evaluation results counters to /metrics","message":"Add evaluation results counters to /metrics\n\nThis change adds code for collection and storage of\nevaluation result counters. The values of the counters can\nthen be retreived through a new /metrics api endpoint.\n\nCounters are collected and served for each alarm separately. Only\nmetrics for the current project will be shown on the /metrics endpoint.\nTo enable this feature set \"enable_metrics\" to true inside [DEFAULT].\n\nChange-Id: I3fceb9a244f248f1e62a1267408c82d1b35a4d37\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/b3fbe13b305f354e7b3b6b2ecd091fefcbb13ede"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/b3fbe13b305f354e7b3b6b2ecd091fefcbb13ede"}]},"branch":"refs/heads/master"},"780bcca10b126725c1863de8c060fa02881a00b1":{"kind":"REWORK","_number":13,"created":"2025-01-16 19:20:41.000000000","uploader":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"ref":"refs/changes/75/925775/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/75/925775/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/75/925775/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/75/925775/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/75/925775/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/75/925775/13"}}},"commit":{"parents":[{"commit":"6f7ba5bbdaff544260966208f1ab54c75deffcca","subject":"Merge \"Simplify tox targets\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/6f7ba5bbdaff544260966208f1ab54c75deffcca"}]}],"author":{"name":"Jaromir Wysoglad","email":"jwysogla@redhat.com","date":"2024-08-06 14:01:56.000000000","tz":-240},"committer":{"name":"Jaromir Wysoglad","email":"jwysogla@redhat.com","date":"2025-01-16 19:15:13.000000000","tz":-300},"subject":"Add evaluation results counters to /metrics","message":"Add evaluation results counters to /metrics\n\nThis change adds code for collection and storage of\nevaluation result counters. The values of the counters can\nthen be retreived through a new /metrics api endpoint.\n\nCounters are collected and served for each alarm separately. Only\nmetrics for the current project will be shown on the /metrics endpoint.\nTo enable this feature set \"enable_metrics\" to true inside [DEFAULT].\n\nChange-Id: I3fceb9a244f248f1e62a1267408c82d1b35a4d37\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/780bcca10b126725c1863de8c060fa02881a00b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/780bcca10b126725c1863de8c060fa02881a00b1"}]},"branch":"refs/heads/master"},"9ab5fb2873a7f8b1abc7e0e8284bad877b25132d":{"kind":"REWORK","_number":14,"created":"2025-01-29 08:51:14.000000000","uploader":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"ref":"refs/changes/75/925775/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/75/925775/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/75/925775/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/75/925775/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/75/925775/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/75/925775/14"}}},"commit":{"parents":[{"commit":"6f7ba5bbdaff544260966208f1ab54c75deffcca","subject":"Merge \"Simplify tox targets\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/6f7ba5bbdaff544260966208f1ab54c75deffcca"}]}],"author":{"name":"Jaromir Wysoglad","email":"jwysogla@redhat.com","date":"2024-08-06 14:01:56.000000000","tz":-240},"committer":{"name":"Jaromir Wysoglad","email":"jwysogla@redhat.com","date":"2025-01-29 08:18:42.000000000","tz":-300},"subject":"Add evaluation results counters to /metrics","message":"Add evaluation results counters to /metrics\n\nThis change adds code for collection and storage of\nevaluation result counters. The values of the counters can\nthen be retreived through a new /metrics api endpoint.\n\nCounters are collected and served for each alarm separately. Only\nmetrics for the current project will be shown on the /metrics endpoint.\nTo enable this feature set \"enable_evaluation_results_metrics\"\nto true inside [DEFAULT].\n\nChange-Id: I3fceb9a244f248f1e62a1267408c82d1b35a4d37\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/9ab5fb2873a7f8b1abc7e0e8284bad877b25132d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/9ab5fb2873a7f8b1abc7e0e8284bad877b25132d"}]},"branch":"refs/heads/master"},"4a86af34ad314708c32e4542511de41343a4eabd":{"kind":"REWORK","_number":15,"created":"2025-02-20 15:05:49.000000000","uploader":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"ref":"refs/changes/75/925775/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/75/925775/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/75/925775/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/75/925775/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/75/925775/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/75/925775/15"}}},"commit":{"parents":[{"commit":"60ef9dd1b95d88d68d4f2f8d66fdb051b439cee6","subject":"Merge \"Fix missing nullable constraint\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/60ef9dd1b95d88d68d4f2f8d66fdb051b439cee6"}]}],"author":{"name":"Jaromir Wysoglad","email":"jwysogla@redhat.com","date":"2024-08-06 14:01:56.000000000","tz":-240},"committer":{"name":"Jaromir Wysoglad","email":"jwysogla@redhat.com","date":"2025-02-20 14:57:06.000000000","tz":-300},"subject":"Add evaluation results counters to /metrics","message":"Add evaluation results counters to /metrics\n\nThis change adds code for collection and storage of\nevaluation result counters. The values of the counters can\nthen be retreived through a new /metrics api endpoint.\n\nCounters are collected and served for each alarm separately. Only\nmetrics for the current project will be shown on the /metrics endpoint.\nTo enable this feature set \"enable_evaluation_results_metrics\"\nto true inside [DEFAULT].\n\nChange-Id: I3fceb9a244f248f1e62a1267408c82d1b35a4d37\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/4a86af34ad314708c32e4542511de41343a4eabd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/4a86af34ad314708c32e4542511de41343a4eabd"}]},"branch":"refs/heads/master"},"1643abed422374733b3e73b4dcc701fd51114b84":{"kind":"REWORK","_number":16,"created":"2025-03-04 15:10:41.000000000","uploader":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"ref":"refs/changes/75/925775/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/75/925775/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/75/925775/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/75/925775/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/75/925775/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/75/925775/16"}}},"commit":{"parents":[{"commit":"60ef9dd1b95d88d68d4f2f8d66fdb051b439cee6","subject":"Merge \"Fix missing nullable constraint\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/60ef9dd1b95d88d68d4f2f8d66fdb051b439cee6"}]}],"author":{"name":"Jaromir Wysoglad","email":"jwysogla@redhat.com","date":"2024-08-06 14:01:56.000000000","tz":-240},"committer":{"name":"Jaromir Wysoglad","email":"jwysogla@redhat.com","date":"2025-03-04 14:14:14.000000000","tz":-300},"subject":"Add evaluation results counters to /metrics","message":"Add evaluation results counters to /metrics\n\nThis change adds code for collection and storage of\nevaluation result counters. The values of the counters can\nthen be retreived through a new /metrics api endpoint.\n\nCounters are collected and served for each alarm separately. Only\nmetrics for the current project will be shown on the /metrics endpoint.\nTo enable this feature set \"enable_evaluation_results_metrics\"\nto true inside [DEFAULT].\n\nChange-Id: I3fceb9a244f248f1e62a1267408c82d1b35a4d37\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/1643abed422374733b3e73b4dcc701fd51114b84"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/1643abed422374733b3e73b4dcc701fd51114b84"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
