)]}'
{"id":"openstack%2Faodh~931196","triplet_id":"openstack%2Faodh~master~Ie5a6c872a16781b25c569667f7b84183b31a1283","project":"openstack/aodh","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-10-03 09:09:15.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ie5a6c872a16781b25c569667f7b84183b31a1283","subject":"Always load api-paste.ini from config directory","status":"ABANDONED","created":"2024-10-02 15:34:04.000000000","updated":"2024-10-03 09:09:15.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"01c471643439a6294b6bb5dd4f1db576e12e2f9e","_number":931196,"virtual_id_number":931196,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"all":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-02 16:44:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e2900b8d8425fbbf988e99d09aadb8dbc611e67f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-02 15:34:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"902bc727dc2fe683a426e851c6e5487c5b8066e9","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-02 16:41:50.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"d6a3d2eb2897c44d5f2830ba5176304be841f389","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-02 16:44:49.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ed63ebb0ecbc4f7c85f36aac971bb5a8\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/6fb8b52ec4f8479ea742a459e788a214 : SUCCESS in 2m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/65f2ac53b81f4052b0b40cfe3bede65f : SUCCESS in 4m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4c6af99da7114cd398f7a9a31c916052 : FAILURE in 4m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/23113ae115bc4a8ba358bfa610b64338 : FAILURE in 7m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e59d412369764fd7aceed156c591420e : SUCCESS in 6m 08s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/e7f798e6718343b88481c4535a4776dc : FAILURE in 37m 56s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/ac13c80aa69241338e87036ec3d03a57 : FAILURE in 19m 59s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/428cfb2c91d142c993ebd20d1d9dc043 : FAILURE in 38m 52s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/112751f6154143de9241d53b3c0a4c78 : FAILURE in 37m 59s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/4cd6b731f19a41a49ba5d2e91fcd5844 : FAILURE in 57m 21s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/acf167fa7713492f907676569ecb9a5e : FAILURE in 58m 16s","accounts_in_message":[],"_revision_number":1},{"id":"64c9a60d07d7411b21a9a085a8202ffc718ba789","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-03 07:04:34.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"ee19ff0c6210fd6d50718e273a520bb01fd39198","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-03 07:05:14.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"89c8f2c8f6ebd3f5a46e18014396ddb119779305","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-03 07:11:33.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"e6212da8be153175bef6dccdcbff70efa7cd6942","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-03 08:16:37.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"09ad72d79e259c94064d89ab53194a6ab6da9ec0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-03 08:19:23.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"8c4eb9b6f25c8450175ac47c23c5a6b475150768","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-03 08:22:37.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"01c471643439a6294b6bb5dd4f1db576e12e2f9e","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-03 09:09:15.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"f82a6eb0fa6c8f22ef9a0d7288c557d32fb805e5","revisions":{"1cdea66d5503c7f0a7fe9bfaca2b8eecff68900b":{"kind":"REWORK","_number":1,"created":"2024-10-02 15:34:04.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/96/931196/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/96/931196/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/96/931196/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/96/931196/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/96/931196/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/96/931196/1"}}},"commit":{"parents":[{"commit":"93f81277f4b41cce71b71695edf8ac14fa67c784","subject":"Merge \"Update master for stable/2024.2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/93f81277f4b41cce71b71695edf8ac14fa67c784"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-02 15:33:08.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-02 15:33:08.000000000","tz":540},"subject":"Always load api-paste.ini from config directory","message":"Always load api-paste.ini from config directory\n\nThe file is not a python code but is actually a config file, so it\nshould be placed in /etc .\n\nChange-Id: Ie5a6c872a16781b25c569667f7b84183b31a1283\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/1cdea66d5503c7f0a7fe9bfaca2b8eecff68900b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/1cdea66d5503c7f0a7fe9bfaca2b8eecff68900b"}]},"branch":"refs/heads/master"},"ba67146200315dbf19e8bc4c8fdcb098c313dd6b":{"kind":"REWORK","_number":2,"created":"2024-10-03 07:04:34.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/96/931196/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/96/931196/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/96/931196/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/96/931196/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/96/931196/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/96/931196/2"}}},"commit":{"parents":[{"commit":"93f81277f4b41cce71b71695edf8ac14fa67c784","subject":"Merge \"Update master for stable/2024.2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/93f81277f4b41cce71b71695edf8ac14fa67c784"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-02 15:33:08.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-03 07:04:20.000000000","tz":540},"subject":"Always load api-paste.ini from config directory","message":"Always load api-paste.ini from config directory\n\nThe file is not a python code but is actually a config file, so it\nshould be placed in /etc .\n\nChange-Id: Ie5a6c872a16781b25c569667f7b84183b31a1283\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/ba67146200315dbf19e8bc4c8fdcb098c313dd6b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/ba67146200315dbf19e8bc4c8fdcb098c313dd6b"}]},"branch":"refs/heads/master"},"565601ff2aa0132057f84ce725e5c6c514d43ac5":{"kind":"REWORK","_number":3,"created":"2024-10-03 07:05:14.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/96/931196/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/96/931196/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/96/931196/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/96/931196/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/96/931196/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/96/931196/3"}}},"commit":{"parents":[{"commit":"93f81277f4b41cce71b71695edf8ac14fa67c784","subject":"Merge \"Update master for stable/2024.2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/93f81277f4b41cce71b71695edf8ac14fa67c784"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-02 15:33:08.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-03 07:05:05.000000000","tz":540},"subject":"Always load api-paste.ini from config directory","message":"Always load api-paste.ini from config directory\n\nThe file is not a python code but is actually a config file, so it\nshould be placed in /etc .\n\nChange-Id: Ie5a6c872a16781b25c569667f7b84183b31a1283\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/565601ff2aa0132057f84ce725e5c6c514d43ac5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/565601ff2aa0132057f84ce725e5c6c514d43ac5"}]},"branch":"refs/heads/master"},"1ee48ec52e3447a279d40a7dc2dfd75cc0876f63":{"kind":"REWORK","_number":4,"created":"2024-10-03 07:11:33.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/96/931196/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/96/931196/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/96/931196/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/96/931196/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/96/931196/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/96/931196/4"}}},"commit":{"parents":[{"commit":"93f81277f4b41cce71b71695edf8ac14fa67c784","subject":"Merge \"Update master for stable/2024.2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/93f81277f4b41cce71b71695edf8ac14fa67c784"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-02 15:33:08.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-03 07:11:02.000000000","tz":540},"subject":"Always load api-paste.ini from config directory","message":"Always load api-paste.ini from config directory\n\nThe file is not a python code but is actually a config file, so it\nshould be placed in /etc .\n\nChange-Id: Ie5a6c872a16781b25c569667f7b84183b31a1283\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/1ee48ec52e3447a279d40a7dc2dfd75cc0876f63"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/1ee48ec52e3447a279d40a7dc2dfd75cc0876f63"}]},"branch":"refs/heads/master"},"5bb8dae0c67c11882f4e879bdba080e6fa222612":{"kind":"REWORK","_number":5,"created":"2024-10-03 08:16:37.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/96/931196/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/96/931196/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/96/931196/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/96/931196/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/96/931196/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/96/931196/5"}}},"commit":{"parents":[{"commit":"93f81277f4b41cce71b71695edf8ac14fa67c784","subject":"Merge \"Update master for stable/2024.2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/93f81277f4b41cce71b71695edf8ac14fa67c784"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-02 15:33:08.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-03 08:16:25.000000000","tz":540},"subject":"Always load api-paste.ini from config directory","message":"Always load api-paste.ini from config directory\n\nThe file is not a python code but is actually a config file, so it\nshould be placed in /etc .\n\nChange-Id: Ie5a6c872a16781b25c569667f7b84183b31a1283\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/5bb8dae0c67c11882f4e879bdba080e6fa222612"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/5bb8dae0c67c11882f4e879bdba080e6fa222612"}]},"branch":"refs/heads/master"},"bd295e8c1748710929974b22496ec228f9e1bdb7":{"kind":"REWORK","_number":6,"created":"2024-10-03 08:19:23.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/96/931196/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/96/931196/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/96/931196/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/96/931196/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/96/931196/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/96/931196/6"}}},"commit":{"parents":[{"commit":"93f81277f4b41cce71b71695edf8ac14fa67c784","subject":"Merge \"Update master for stable/2024.2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/93f81277f4b41cce71b71695edf8ac14fa67c784"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-02 15:33:08.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-03 08:19:14.000000000","tz":540},"subject":"Always load api-paste.ini from config directory","message":"Always load api-paste.ini from config directory\n\nThe file is not a python code but is actually a config file, so it\nshould be placed in /etc .\n\nChange-Id: Ie5a6c872a16781b25c569667f7b84183b31a1283\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/bd295e8c1748710929974b22496ec228f9e1bdb7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/bd295e8c1748710929974b22496ec228f9e1bdb7"}]},"branch":"refs/heads/master"},"f82a6eb0fa6c8f22ef9a0d7288c557d32fb805e5":{"kind":"REWORK","_number":7,"created":"2024-10-03 08:22:37.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/96/931196/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/96/931196/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/96/931196/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/96/931196/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/96/931196/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/96/931196/7"}}},"commit":{"parents":[{"commit":"93f81277f4b41cce71b71695edf8ac14fa67c784","subject":"Merge \"Update master for stable/2024.2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/93f81277f4b41cce71b71695edf8ac14fa67c784"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-02 15:33:08.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-03 08:22:28.000000000","tz":540},"subject":"Always load api-paste.ini from config directory","message":"Always load api-paste.ini from config directory\n\nThe file is not a python code but is actually a config file, so it\nshould be placed in /etc .\n\nChange-Id: Ie5a6c872a16781b25c569667f7b84183b31a1283\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/f82a6eb0fa6c8f22ef9a0d7288c557d32fb805e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/f82a6eb0fa6c8f22ef9a0d7288c557d32fb805e5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
