)]}'
{"id":"openstack%2Faodh~931881","triplet_id":"openstack%2Faodh~master~I92bd612dae86bb6a8a1342daee54bdf00098501c","project":"openstack/aodh","branch":"master","attention_set":{},"removed_from_attention_set":{"32240":{"account":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},"last_update":"2024-10-14 12:47:02.000000000","reason":"Change was submitted"},"5202":{"account":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"last_update":"2024-10-14 11:48:08.000000000","reason":"\u003cGERRIT_ACCOUNT_5202\u003e replied on the change","reason_account":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"}},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-10-14 12:47:02.000000000","reason":"Change was submitted"},"4264":{"account":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"last_update":"2024-10-14 06:26:06.000000000","reason":"\u003cGERRIT_ACCOUNT_4264\u003e replied on the change","reason_account":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"}}},"hashtags":[],"change_id":"I92bd612dae86bb6a8a1342daee54bdf00098501c","subject":"Make PostgreSQL functional job non-voting","status":"MERGED","created":"2024-10-09 07:58:10.000000000","updated":"2024-10-14 12:51:15.000000000","submitted":"2024-10-14 12:47:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"931881","meta_rev_id":"cb11ad12d48ee2a338812a450c743906b2b8f92a","_number":931881,"virtual_id_number":931881,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-10-14 12:47:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"all":[{"value":2,"date":"2024-10-14 11:48:08.000000000","_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":2,"date":"2024-10-14 06:26:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"all":[{"value":1,"date":"2024-10-14 11:48:08.000000000","_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"}],"reviewers":{"REVIEWER":[{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-09 08:56:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-10 05:55:38.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"},{"updated":"2024-10-10 05:55:38.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},"state":"REVIEWER"},{"updated":"2024-10-10 05:55:38.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"state":"REVIEWER"}],"messages":[{"id":"d4a2b69507f89f40342999529b922dc5e23b4692","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-09 07:58:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9c982fe57a38e4a166665be11314b5d3ff83dbe8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-09 08:56:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/23cdfafbdab64bf0915b53e2a34cced7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0455ca20ed85415dbb9d9d2c9bfd128e : SUCCESS in 3m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bb7dd127a99f48228d2d64e8451f0fd3 : SUCCESS in 4m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/00c0f33129374bb599f053d65f84ad44 : SUCCESS in 6m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c5658dd5d6e4069b5a6ea4d71ee8fae : SUCCESS in 5m 59s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/bc8990dfae914be89eddbe7546d07f7c : SUCCESS in 28m 56s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/5f8651dda2614a2abc14b6f4e9dce29c : FAILURE in 30m 44s (non-voting)\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/34d575b877794a0d9cad99a12b82d992 : SUCCESS in 35m 36s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/3b2944484569421e801176dda3538230 : SUCCESS in 33m 29s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/39936b0728444d25a2da72db8250e093 : SUCCESS in 39m 54s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/d7e6f85107b343ba935545587d8e84fc : SUCCESS in 56m 44s","accounts_in_message":[],"_revision_number":1},{"id":"f1b3d3b5e4321d1be2b880f560bcaa26dde91345","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-10 05:55:38.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"19f4b77df2116c04b4a87f094cc1219fbdb9b65d","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2024-10-14 06:26:06.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bb67eb15b7fed9cfdcd528efb73094a84c8a3c11","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-14 06:30:46.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ab1825bebe6d667cf0afb96b76c1c7c2bdffc9ca","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2024-10-14 11:48:08.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"99c133cc43b3d4c439fa5e8420161d94517b9e6d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-14 11:49:02.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6bebcc2dda3a4a167c5cba76a79a82f8c37565cb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-14 12:47:02.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bc703b91104948aa89ee8af98197467f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b41739e6785b4b53abc53666624abc7e : SUCCESS in 3m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/be7bb0d4c23140b3acb1e3648540d2b8 : SUCCESS in 4m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f69c625fa686488ab1b247ae975083be : SUCCESS in 5m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f2be810ae2e467f877094ab2f881f08 : SUCCESS in 6m 30s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/ef9288219bcd4524b0d696e02c2b128d : SUCCESS in 35m 48s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/a656636411c342028c37f225ac195cff : FAILURE in 26m 57s (non-voting)\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/970c9d9b014a48d59ad2daefd1971a00 : SUCCESS in 50m 31s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/50683f210fa048e49503759ee8ed1e28 : SUCCESS in 50m 30s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/a628fe3fb45d43e3a04915bdc5a3eaf9 : SUCCESS in 48m 02s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/048d3b02d0dd4af6938b93ef2846c0aa : SUCCESS in 44m 31s","accounts_in_message":[],"_revision_number":1},{"id":"8b7084bc98437e30c78cddcaeeb3af7ff59d76ea","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-14 12:47:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"cb11ad12d48ee2a338812a450c743906b2b8f92a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-14 12:51:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b0dff6d980a34cd3addf02bb7c432b32\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/394aaf0d1a7d42aba36527d10107b8a2 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e52699f9eb36a8c761159659ed8a510f66f3ab19","revisions":{"e52699f9eb36a8c761159659ed8a510f66f3ab19":{"kind":"REWORK","_number":1,"created":"2024-10-09 07:58:10.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/81/931881/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/81/931881/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/81/931881/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/81/931881/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/81/931881/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/81/931881/1"}}},"commit":{"parents":[{"commit":"6f7ba5bbdaff544260966208f1ab54c75deffcca","subject":"Merge \"Simplify tox targets\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/6f7ba5bbdaff544260966208f1ab54c75deffcca"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-09 07:56:37.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-09 07:56:37.000000000","tz":540},"subject":"Make PostgreSQL functional job non-voting","message":"Make PostgreSQL functional job non-voting\n\nThis job has been broken because of a known regression in openstack\nCLI. The issue is being fixed in 7.1.3 but the release is not yet\npulled into master. Make the job non-voting to unblock the gate now.\n\nChange-Id: I92bd612dae86bb6a8a1342daee54bdf00098501c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/e52699f9eb36a8c761159659ed8a510f66f3ab19"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/e52699f9eb36a8c761159659ed8a510f66f3ab19"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
