)]}'
{"id":"openstack%2Faodh~932233","triplet_id":"openstack%2Faodh~master~I8424baa33a837d7e52a04384050660dab2977ea9","project":"openstack/aodh","branch":"master","topic":"use_metadata_in_autoscaling","attention_set":{},"removed_from_attention_set":{"34975":{"account":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"last_update":"2024-10-16 18:38:29.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I8424baa33a837d7e52a04384050660dab2977ea9","subject":"Use server grouping by metadata in autoscaling","status":"ABANDONED","created":"2024-10-11 20:41:43.000000000","updated":"2024-10-16 18:38:29.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"9eceec2be6bd16b8e49c5348cd961579b42d1ffc","_number":932233,"virtual_id_number":932233,"owner":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"tag":"autogenerated:zuul:check","value":1,"date":"2024-10-15 15:23:21.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"all":[{"value":-1,"date":"2024-10-16 10:40:42.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":2,"date":"2024-10-16 07:03:35.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-11 21:16:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-16 07:03:35.000000000","updated_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"reviewer":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"state":"REVIEWER"},{"updated":"2024-10-16 10:40:42.000000000","updated_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"}],"messages":[{"id":"efc709d02e5565798166749674ca43e578c9a166","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2024-10-11 20:41:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"757cb10cd5e07d1d0a97b9a470427879a6990435","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-11 21:16:27.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d079d99b1d964309a8bf81b132e8d088\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe18400beec64e57aab1b87fd04f68f9 : SUCCESS in 3m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/27c1c2c345e64dca909459c982d8b5af : SUCCESS in 4m 08s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d76d32a39d884a45bc889630a1f6f304 : SUCCESS in 4m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/768414351d9b46868c70425351fe16e9 : SUCCESS in 5m 32s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/4c82b8aeb85a44b9bc696d7614772b9c : SUCCESS in 22m 13s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/7998e03adc2c4d578853912c4b8de938 : FAILURE in 14m 09s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/93f9aba0e0fa435bb3796771b7f1553f : SUCCESS in 29m 05s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/e8b8030318904eb5a1be58b9ddf1508d : SUCCESS in 27m 44s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/f063ba652b3c45de99206ae3241890c1 : SUCCESS in 29m 14s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/c4d249ef618d46cf933ef5911145a389 : SUCCESS in 33m 36s","accounts_in_message":[],"_revision_number":1},{"id":"6db20a8f95201659bd233a52ea79757f283e4a03","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2024-10-15 14:21:08.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5c0bb240351f23328953d5f60df7325510c19ced","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-15 15:23:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b60d6ed22a394050bff970e95b43fec8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ff04254dca74ab5a84d80b51b83627a : SUCCESS in 3m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4e25dc952ef943329bb5367589ff8415 : SUCCESS in 4m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/aba4dbdbb0cc421697d50e81c2bb21af : SUCCESS in 7m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1289170708d64ce29fe296dea3d5125b : SUCCESS in 6m 17s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/6811366df43b489ea60938941859aab0 : SUCCESS in 31m 23s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/a06d73eaa29243618a7096f0edd86d7c : FAILURE in 28m 32s (non-voting)\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/a59fd277c69445e9a45511e0d042dc50 : SUCCESS in 43m 06s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/3e980bff7f9e45eeadd1d6e2633b2eeb : SUCCESS in 33m 53s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/7ad0166fbad1459a8cce3db52ee6fa5f : SUCCESS in 58m 15s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/c473a7b2594849d89d98975dcc05bca1 : SUCCESS in 57m 04s","accounts_in_message":[],"_revision_number":1},{"id":"2d3425495792911646b6a8d3dad37279ae863eb2","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2024-10-16 07:03:35.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"207bfdc2f18b14e2acc18194d032e2987be0c2eb","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2024-10-16 10:40:42.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9eceec2be6bd16b8e49c5348cd961579b42d1ffc","tag":"autogenerated:gerrit:abandon","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2024-10-16 18:38:29.000000000","message":"Abandoned\n\nAbandoned in favor of https://review.opendev.org/c/openstack/telemetry-tempest-plugin/+/932548","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"34d0e00831b73256e74401876ed92ff210acc62c","revisions":{"34d0e00831b73256e74401876ed92ff210acc62c":{"kind":"REWORK","_number":1,"created":"2024-10-11 20:41:43.000000000","uploader":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"ref":"refs/changes/33/932233/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/33/932233/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/33/932233/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/33/932233/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/33/932233/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/33/932233/1"}}},"commit":{"parents":[{"commit":"6f7ba5bbdaff544260966208f1ab54c75deffcca","subject":"Merge \"Simplify tox targets\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/6f7ba5bbdaff544260966208f1ab54c75deffcca"}]}],"author":{"name":"Jaromir Wysoglad","email":"jwysogla@redhat.com","date":"2024-10-11 20:41:19.000000000","tz":-240},"committer":{"name":"Jaromir Wysoglad","email":"jwysogla@redhat.com","date":"2024-10-11 20:41:41.000000000","tz":-240},"subject":"Use server grouping by metadata in autoscaling","message":"Use server grouping by metadata in autoscaling\n\nChange-Id: I8424baa33a837d7e52a04384050660dab2977ea9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/34d0e00831b73256e74401876ed92ff210acc62c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/34d0e00831b73256e74401876ed92ff210acc62c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
