)]}'
{"id":"openstack%2Faodh~932550","triplet_id":"openstack%2Faodh~stable%2F2023.2~I6eac76234b318caa90cceec192e7c31b81f984d2","project":"openstack/aodh","branch":"stable/2023.2","attention_set":{},"removed_from_attention_set":{"34975":{"account":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"last_update":"2024-10-30 16:40:29.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I6eac76234b318caa90cceec192e7c31b81f984d2","subject":"[DNM] Testing CI on stable branches","status":"ABANDONED","created":"2024-10-16 18:50:22.000000000","updated":"2024-10-30 16:40:29.000000000","total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"446264dbef7f7ff663c6b68052cda9e01498380a","_number":932550,"virtual_id_number":932550,"owner":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-10-16 19:52:38.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-16 19:52:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"13eaad0ec505ca698b97d88e3959334aa5e570d6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2024-10-16 18:50:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5b589ecaaf6f8a92d9a3b3913149372237a6e1eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-16 19:52:38.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cb0443a2c4e34c77be62cc23ff3ce461\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df54a0df1b624819949a707e4198bf42 : SUCCESS in 3m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5cf620b4637a4bdd82cb27b0c86b95f8 : SUCCESS in 6m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ad0922a6e410427798fdc8c4f5fa34cb : SUCCESS in 4m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e349f0b150774e9bbcc5ec27318e1cdf : SUCCESS in 4m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4a5d0b5cb7a441a1b83df1d68689b419 : RETRY_LIMIT in 1m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b82ddd071d6f408b82893571857e09d1 : SUCCESS in 5m 43s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/2c0d9237f6534dbc9536dfed6a991c05 : SUCCESS in 44m 14s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/5fd2a2e4523249c99661217c63508ebe : FAILURE in 45m 31s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/b3aaaccf6ff145e9a84fcbcc024ccaa4 : FAILURE in 51m 13s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/8859500b14eb42129ca006d179d07035 : FAILURE in 53m 34s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/ed02f24d2584474baf45d40020940c27 : FAILURE in 56m 23s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/5138e335f6ab4e84be27e49a431efaad : FAILURE in 56m 55s","accounts_in_message":[],"_revision_number":1},{"id":"44b270ce15ff9114518023d03914238cba9dc7e8","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2024-10-16 22:05:18.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cd0c6f78531335e437fab51055f7ceb66a3e93ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-16 23:11:34.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/043771bf0a0141bbbc0267a9abb888c2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/096e5ebdc3f54a248fc78c53be95174b : SUCCESS in 3m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/97d7e5e740bb44c78ab6b41d0cc6a901 : SUCCESS in 6m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/63c8985039314e399f3c005652fc71b7 : SUCCESS in 6m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/aae67400e2a74a3dbcea14d04a7c1111 : FAILURE in 5m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/17df66020bfb40729d4e855bc7b0f743 : RETRY_LIMIT in 3m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/afb81ed198c24e3eb27cf069ff705be9 : SUCCESS in 5m 57s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/10faeefc900748f28345d191bf2c052d : SUCCESS in 41m 20s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/ac24b702015b4f72970dcf6d33d476b1 : FAILURE in 42m 46s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/1245bf2846a442dabc7e0395900a6259 : FAILURE in 46m 38s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/72aa44179eca4e77a6a845acc6f80d40 : FAILURE in 47m 23s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/0fc7cb03e5984e24bb37ad3558cd56d7 : FAILURE in 59m 53s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/05b9121abdc54158b25ff3e1d676f502 : FAILURE in 1h 00m 47s","accounts_in_message":[],"_revision_number":1},{"id":"5f70c37af115134fb0ca8d2ce99115f021a2d7e8","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2024-10-17 08:36:13.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"31fabd85cdd40706e22f3c6b306ac1532d50ac2b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-17 09:42:55.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2a85e72fcb0b4563b7762e03eb533068\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a41318f5623d42fbbe2d012ad8ae8f19 : SUCCESS in 2m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/96f6fa508f7942e7a00409f8e3c15112 : SUCCESS in 9m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f1df9c0359884708bbe57dedd929643c : SUCCESS in 4m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/94c788186f804efeb4c6d12852fb27d2 : SUCCESS in 5m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bb020c38a8db4a7a9e704c74f4723215 : RETRY_LIMIT in 1m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d3f935fd52b49b4ae8edc01919a302d : SUCCESS in 5m 40s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/00adf2cef931419c9373183b00570437 : SUCCESS in 37m 00s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/c9a7f75320c645ea8990b45f64079429 : FAILURE in 30m 43s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/e4516bf9b25c40858b67e0699a2a5334 : SUCCESS in 48m 12s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/3ef4f85c1ffd46dc908ab8d5ef3a083d : FAILURE in 34m 20s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/d6bae6a582814f4291e96375c82a322b : SUCCESS in 38m 28s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/ebd0d6e0e44d42e0a263a4560d53f3f9 : SUCCESS in 1h 00m 59s","accounts_in_message":[],"_revision_number":1},{"id":"ab55b468df2f0094bf4363406838dae73dfbeb27","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2024-10-17 09:59:51.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1702eb1f792519ad0bda6204b7325439d170d0ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-17 10:56:16.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/911f8ade16dd40ae83c33925134e2782\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc7eb0d9317f4c1a8441e4765d4ccb15 : SUCCESS in 4m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c7f7d74089b24d4fb5758e5694c8f866 : SUCCESS in 8m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/def3aa67e3a94fe7a1d80e9fdfaf204e : SUCCESS in 7m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/81cbeea01ee74f71a44a11478be85c06 : SUCCESS in 7m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4dc67ea6ecda46ddbcdf4b4a4cb2df95 : RETRY_LIMIT in 1m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e084af405d24387afd3854dbb90ee36 : SUCCESS in 6m 19s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/4f4f4915d04c4bdcab3c007258d3d7c6 : SUCCESS in 44m 47s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/3f912be3606d42f4ba4fa2a3aa880ae0 : FAILURE in 23m 54s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/72d69a0a47dc4bc2a4e8ba7f087fff53 : SUCCESS in 55m 24s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/6d00bbb3410e47acb50f921065843d67 : SUCCESS in 27m 07s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/7f102b874fcf41fcbaa9dbd66a630e6c : FAILURE in 53m 23s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/4b6a592436d649c3a169c709e83996c6 : SUCCESS in 34m 44s","accounts_in_message":[],"_revision_number":1},{"id":"446264dbef7f7ff663c6b68052cda9e01498380a","tag":"autogenerated:gerrit:abandon","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2024-10-30 16:40:29.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"507cd709471d7a7d9dfaf6258b7138ec83190da7","revisions":{"507cd709471d7a7d9dfaf6258b7138ec83190da7":{"kind":"REWORK","_number":1,"created":"2024-10-16 18:50:22.000000000","uploader":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"ref":"refs/changes/50/932550/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/50/932550/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/50/932550/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/50/932550/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/50/932550/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/50/932550/1"}}},"commit":{"parents":[{"commit":"1215c3fb4aaf8a9d7ab7ef91dacf40120c64f371","subject":"Merge \"Update TOX_CONSTRAINTS_FILE for stable/2023.2\" into stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/1215c3fb4aaf8a9d7ab7ef91dacf40120c64f371"}]}],"author":{"name":"Jaromir Wysoglad","email":"jwysogla@redhat.com","date":"2024-10-16 18:48:19.000000000","tz":-240},"committer":{"name":"Jaromir Wysoglad","email":"jwysogla@redhat.com","date":"2024-10-16 18:48:19.000000000","tz":-240},"subject":"[DNM] Testing CI on stable branches","message":"[DNM] Testing CI on stable branches\n\nDepends-On: https://review.opendev.org/c/openstack/telemetry-tempest-plugin/+/932548\nChange-Id: I6eac76234b318caa90cceec192e7c31b81f984d2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/507cd709471d7a7d9dfaf6258b7138ec83190da7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/507cd709471d7a7d9dfaf6258b7138ec83190da7"}]},"branch":"refs/heads/stable/2023.2"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
