)]}'
{"id":"openstack%2Faodh~932578","triplet_id":"openstack%2Faodh~master~Ia05b31061485e613634b27e96b904ad3ebe8fa33","project":"openstack/aodh","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-11-15 16:17:00.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"Ia05b31061485e613634b27e96b904ad3ebe8fa33","subject":"Add note about lower bound of requirements","status":"MERGED","created":"2024-10-17 02:04:05.000000000","updated":"2024-11-15 16:31:27.000000000","submitted":"2024-11-15 16:30:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"932578","meta_rev_id":"3a06c60087c3eccc3a948fe991420629c012d95c","_number":932578,"virtual_id_number":932578,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-15 16:30:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"all":[{"value":2,"date":"2024-10-30 08:16:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":2,"date":"2024-11-15 16:17:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":1,"date":"2024-11-15 16:17:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-17 02:39:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-30 08:16:25.000000000","updated_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"reviewer":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"state":"REVIEWER"}],"messages":[{"id":"b89c7dfd79e1a580106fa73252ba43cb43afd109","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-17 02:04:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3c1ee3fae3bb0bdc605a930f8f9b0f2c64df71c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-17 02:39:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cabb90bf5f5d4c75b226531739fb853e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a3a373e26dec47d6a8eef09e867a9502 : SUCCESS in 4m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4774737469d5469a9a78e390b156c4e7 : SUCCESS in 4m 40s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a2668d2404d24ca895026c4bf0d30ea0 : SUCCESS in 5m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd6df8033c08413d9f8733acd384d799 : SUCCESS in 7m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c21447e4bd7945fbaa7d2ed56196b0f8 : SUCCESS in 4m 57s\n- requirements-check https://zuul.opendev.org/t/openstack/build/aaf1e96862f14684862bad46ea0d21ff : SUCCESS in 3m 54s","accounts_in_message":[],"_revision_number":1},{"id":"c56f89d8aac069c5661256dec3c3066a40132f46","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2024-10-30 08:16:25.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f91420908684391aa675b5ed7823dcba3b5fdfbc","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-15 16:17:00.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f080ef5176a985c2bbe82b92d1874dc2a9154428","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-15 16:17:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"074ea0a833f04c67e9958f8b8b26b1fc1b2a9269","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-15 16:30:00.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ec1273ca9bc042a3bf25ccafe559a29c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e29623784796488fac1b15f7c525962a : SUCCESS in 2m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f940d47dcad341ec926fa7d6628c7417 : SUCCESS in 3m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f3de8a12b995424bbf9f3e60faf6430b : SUCCESS in 4m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71c70bc0789d4cf4be702bbb9446bb1b : SUCCESS in 7m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/80d071df06d84dfcb88c321c83cdf216 : SUCCESS in 5m 33s\n- requirements-check https://zuul.opendev.org/t/openstack/build/70d8b51565e34781aa0b3cf9ccf49297 : SUCCESS in 2m 49s","accounts_in_message":[],"_revision_number":1},{"id":"6ceec18595a682d0965a00c2a54baf289fb5d8d2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-15 16:30:00.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"3a06c60087c3eccc3a948fe991420629c012d95c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-15 16:31:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/04f73c23840d4143b6ed69d640ee203c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16cf157e1c2c4055be69bf1acb58b237 : SUCCESS in 49s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/70020ef320ae413f8fa4e14f0055225e : SUCCESS in 59s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ca2401aa8d093e42c0ac297cc52f19ba3df19dd2","revisions":{"ca2401aa8d093e42c0ac297cc52f19ba3df19dd2":{"kind":"REWORK","_number":1,"created":"2024-10-17 02:04:05.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/78/932578/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/78/932578/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/78/932578/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/78/932578/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/78/932578/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/78/932578/1"}}},"commit":{"parents":[{"commit":"ebe6170fbc5abfb4a25ceb85e14d41b90b5a28dd","subject":"Merge \"Switch back to built-in md5 function\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/ebe6170fbc5abfb4a25ceb85e14d41b90b5a28dd"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-17 02:02:00.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-17 02:03:55.000000000","tz":540},"subject":"Add note about lower bound of requirements","message":"Add note about lower bound of requirements\n\nSince we removed lower-constrants job, the lower bound is no longer\ntested. Import the note from the other project to declare that it is\nmaintained on bast effort basis.\n\nAlso add license information to all libraries .\n\nChange-Id: Ia05b31061485e613634b27e96b904ad3ebe8fa33\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/ca2401aa8d093e42c0ac297cc52f19ba3df19dd2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/ca2401aa8d093e42c0ac297cc52f19ba3df19dd2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
