)]}'
{"id":"openstack%2Faodh~932590","triplet_id":"openstack%2Faodh~stable%2F2024.1~Id42f1fe9ab4a534fc0b1c74ba0b0de1515dee528","project":"openstack/aodh","branch":"stable/2024.1","hashtags":[],"change_id":"Id42f1fe9ab4a534fc0b1c74ba0b0de1515dee528","subject":"[DNM] Testing CI on stable branches","status":"ABANDONED","created":"2024-10-17 08:36:34.000000000","updated":"2024-10-30 16:40:41.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"96df10b155bf90d08bcee49fcff8df84ee491b65","_number":932590,"virtual_id_number":932590,"owner":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2024-10-17 09:48:31.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-17 09:48:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5933ce7cb99c734547777476e29f6d33f43a9fea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2024-10-17 08:36:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"35c373f8793f1f54725578dbc93db2bc05c76017","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-17 09:48:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e98ec501195c4356a508847da82f9a42\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/77b5de2b94c2410194aec5cc6af40835 : SUCCESS in 4m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1db11bf0289e49618a1c71f50acd48bd : SUCCESS in 6m 28s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9525e326aea34e5cb7b8d02aa0f76ce6 : SUCCESS in 3m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0bdf7ebb5e9544ce8d453d911a903f42 : SUCCESS in 7m 18s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/4bacc8fb6aa9439ea8843fb3814ed27b : SUCCESS in 38m 25s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/c6cc2d59764b4a5f9ebffad61e276be7 : SUCCESS in 30m 35s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/ab15970ca8214c82a3e0f0c41e5db159 : SUCCESS in 38m 25s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/49cd977405a14dfd8291f624d29531fc : SUCCESS in 49m 35s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/b3b06ea571b64e17a97c349a0d4169b7 : SUCCESS in 59m 22s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/1ee39d59e26343afb96965aa6da8fc93 : SUCCESS in 1h 05m 55s","accounts_in_message":[],"_revision_number":1},{"id":"96df10b155bf90d08bcee49fcff8df84ee491b65","tag":"autogenerated:gerrit:abandon","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2024-10-30 16:40:41.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0821119d6c70157dbe4889e9d7e050a291deecfd","revisions":{"0821119d6c70157dbe4889e9d7e050a291deecfd":{"kind":"REWORK","_number":1,"created":"2024-10-17 08:36:34.000000000","uploader":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"ref":"refs/changes/90/932590/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/90/932590/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/90/932590/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/90/932590/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/90/932590/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/90/932590/1"}}},"commit":{"parents":[{"commit":"523062f4aa62640d12f271d952e5308194245c3b","subject":"Update TOX_CONSTRAINTS_FILE for stable/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/523062f4aa62640d12f271d952e5308194245c3b"}]}],"author":{"name":"Jaromir Wysoglad","email":"jwysogla@redhat.com","date":"2024-10-17 08:35:02.000000000","tz":-240},"committer":{"name":"Jaromir Wysoglad","email":"jwysogla@redhat.com","date":"2024-10-17 08:35:02.000000000","tz":-240},"subject":"[DNM] Testing CI on stable branches","message":"[DNM] Testing CI on stable branches\n\nDepends-On: https://review.opendev.org/c/openstack/telemetry-tempest-plugin/+/932548\nChange-Id: Id42f1fe9ab4a534fc0b1c74ba0b0de1515dee528\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/0821119d6c70157dbe4889e9d7e050a291deecfd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/0821119d6c70157dbe4889e9d7e050a291deecfd"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
