)]}'
{"id":"openstack%2Faodh~932591","triplet_id":"openstack%2Faodh~master~I24af0f683962d962207421a77da8d8f5e0936061","project":"openstack/aodh","branch":"master","hashtags":[],"change_id":"I24af0f683962d962207421a77da8d8f5e0936061","subject":"Allow overriding AODH_SERVICE_HOST","status":"MERGED","created":"2024-10-17 08:43:01.000000000","updated":"2024-11-16 17:21:13.000000000","submitted":"2024-11-16 17:20:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"932591","meta_rev_id":"59f2934ddb6a5b0762aa6da7220ab40162255b12","_number":932591,"virtual_id_number":932591,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-16 17:20:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2024-11-16 16:09:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2024-11-16 16:09:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-17 09:38:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ad267cc91bedcf588f33d099af3245c437870055","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-17 08:43:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7f3faa32f36a5c27c49930b854af0ce39f33ff28","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-17 09:38:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4f7786959cb04bfb99e9d2780b85ff73\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5b16633d7bfc4722b7b67eded7c5c394 : SUCCESS in 3m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/07f3c88651bf487abc1c9daf976f3a4f : SUCCESS in 5m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cc53ebb1ad914fb49f32c04d7f1fc4d3 : SUCCESS in 5m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1181e685f2854f2ab30c9245fc099b19 : SUCCESS in 7m 51s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/b133f2f801444816b21c2052e0c909e7 : SUCCESS in 39m 18s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/9c60d07820834d588f74c09f5bc0e6d6 : FAILURE in 20m 28s (non-voting)\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/f2efcfdc9d1c48ab80b76e2c5d182947 : SUCCESS in 47m 12s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/0e6fdf84dd7247f5bc47d4f469013529 : SUCCESS in 33m 16s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/e3920ad864e44127a65d6a18cbd9205a : SUCCESS in 49m 50s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/aeb547add2454671968dc0159fe3595e : SUCCESS in 45m 47s","accounts_in_message":[],"_revision_number":1},{"id":"f046f7a6e47db14e9584812be6f8e1913b14401c","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-16 16:09:50.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ae2e263df65f152edebf4a625c5fca3e4f478891","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-16 16:10:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"661d54530677fd7a40fcbbface13650adaa6f4fb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-16 17:20:17.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c75a462528024ab58a060ecc2535e39c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa384feadb334e05b0b1fceedf19a2a3 : SUCCESS in 3m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/01ce1bd22ec44737bfe114620ea880b8 : SUCCESS in 4m 50s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/426cf9a0eef1487a865db309f3b0ed99 : SUCCESS in 5m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37eb780079444ced918471fa43c092f0 : SUCCESS in 5m 44s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/287abb422ce24894804eba231c418e83 : SUCCESS in 43m 19s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/1c0c054cfc8c4e38ae6e81f6e18b4424 : SUCCESS in 20m 56s (non-voting)\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/0305c87b659c45c89c65f6491bc4d0f4 : SUCCESS in 28m 05s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/9214044af0694199b6889eaa0452f322 : SUCCESS in 52m 45s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/9254808f57aa42afb8e33cf4fad65cf2 : SUCCESS in 53m 47s","accounts_in_message":[],"_revision_number":1},{"id":"5701dabe24e8f628720101aaf0218bd29a058de6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-16 17:20:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"59f2934ddb6a5b0762aa6da7220ab40162255b12","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-16 17:21:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e78909ef6d43493faea56c1e2787e3f0\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1392adf583b543dd810a7d41638d3c55 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"329894210cba39665584cdc4c3e3eddf477a171e","revisions":{"329894210cba39665584cdc4c3e3eddf477a171e":{"kind":"REWORK","_number":1,"created":"2024-10-17 08:43:01.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/91/932591/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/91/932591/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/91/932591/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/91/932591/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/91/932591/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/91/932591/1"}}},"commit":{"parents":[{"commit":"ebe6170fbc5abfb4a25ceb85e14d41b90b5a28dd","subject":"Merge \"Switch back to built-in md5 function\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/ebe6170fbc5abfb4a25ceb85e14d41b90b5a28dd"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-17 08:42:33.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-17 08:42:33.000000000","tz":540},"subject":"Allow overriding AODH_SERVICE_HOST","message":"Allow overriding AODH_SERVICE_HOST\n\n... to use a static host in aodhclient tests.\n\nChange-Id: I24af0f683962d962207421a77da8d8f5e0936061\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/329894210cba39665584cdc4c3e3eddf477a171e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/329894210cba39665584cdc4c3e3eddf477a171e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
