)]}'
{"id":"openstack%2Faodh~934564","triplet_id":"openstack%2Faodh~master~Iba548429fefef53d043d42de4d12554d77e4591a","project":"openstack/aodh","branch":"master","topic":"policy-yaml-file","attention_set":{},"removed_from_attention_set":{"32240":{"account":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},"last_update":"2024-11-18 07:50:29.000000000","reason":"Change was submitted"},"5202":{"account":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"last_update":"2024-11-18 07:50:29.000000000","reason":"Change was submitted"},"4264":{"account":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"last_update":"2024-11-18 06:46:37.000000000","reason":"removed on reply"},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2024-11-18 07:50:29.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iba548429fefef53d043d42de4d12554d77e4591a","subject":"Remove default override for config options policy_file","status":"MERGED","created":"2024-11-11 05:40:26.000000000","updated":"2024-11-18 07:51:21.000000000","submitted":"2024-11-18 07:50:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"934564-policy-yaml-file","meta_rev_id":"932fdd28cde528907ded7f13c8c1b74dd480e56b","_number":934564,"virtual_id_number":934564,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-18 07:50:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2024-11-11 07:08:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2024-11-18 06:46:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2024-11-18 06:46:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"}],"reviewers":{"REVIEWER":[{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-11 06:46:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-11 07:08:34.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2024-11-16 03:16:49.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"state":"REVIEWER"},{"updated":"2024-11-16 03:26:02.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"},{"updated":"2024-11-16 03:26:02.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},"state":"REVIEWER"}],"messages":[{"id":"5f31cbec29fd45b364b5a51c176b21c1a449f1a5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-11-11 05:40:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3b1d7f1a75b31ba1ee31a2f469af09168fb9c617","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-11-11 05:55:19.000000000","message":"Topic policy-json-file removed","accounts_in_message":[],"_revision_number":1},{"id":"48fdc275bc23fd7ec54a107769f474b872f3b87a","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-11-11 05:55:24.000000000","message":"Topic set to policy-yaml-file","accounts_in_message":[],"_revision_number":1},{"id":"8b625714007a6e1c545d1439353e934ea6362935","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-11 06:46:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bd570e280c6f463fa19c9d4b0a4224f1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bcb389d4512441398613d6edf4db281c : SUCCESS in 3m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e876c0ec908a49049583aefcfd7a8ac2 : SUCCESS in 5m 39s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/34f81112afe84df9aeb1f43f5554f3c7 : SUCCESS in 6m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3faca71f62fc480eb7366a9f7edce0d1 : SUCCESS in 6m 59s\n- requirements-check https://zuul.opendev.org/t/openstack/build/394ca5db74c94ecd99264c300b290d15 : SUCCESS in 2m 03s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/3734fc2ff8e14b80a274decbdd2d1b79 : SUCCESS in 40m 23s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/faf63906695c4eba96643e821b7c1994 : SUCCESS in 23m 25s (non-voting)\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/34a56dd06a04483ea9ae7d143e274847 : SUCCESS in 38m 14s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/74a76c56c2af4281bd99ffbd95062117 : SUCCESS in 36m 32s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/418c6fcf189f4d6fb0fd58c45133db7a : SUCCESS in 49m 18s","accounts_in_message":[],"_revision_number":1},{"id":"32a2df23f69c5d4356bf98aabc3eafd4a2ad1793","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-11 07:08:34.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4b142336f9ec6061542432a7220264b192a60ef0","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2024-11-18 06:46:37.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b99b31bada62c99fcf4837698d285b108f614387","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2024-11-18 06:46:43.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0f118ed4e52bb5e6c4653dc3e000b5a6b4f5df76","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-18 06:47:15.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fd059877fb6d8bb6975ef18118d538f338f85760","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-18 07:50:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/116c09ca06744620a4879ef8c5c94f9b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fdcdcc2212564edbadac778f19ab6201 : SUCCESS in 3m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/22e4cfd194344ad89c02703e9c1d08ee : SUCCESS in 6m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0c958cb7d976472ab0e5f275c756d831 : SUCCESS in 5m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1eedabc5dcbd44e0ba1e8a7be2d94056 : SUCCESS in 6m 36s\n- requirements-check https://zuul.opendev.org/t/openstack/build/ca23de2294aa42a6a94726a777edd7fa : SUCCESS in 2m 06s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/1a4b702d57d54d259113b3b1353b039c : SUCCESS in 36m 08s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/67c5dd2e19ee4787af36f0be80e8a8aa : SUCCESS in 19m 43s (non-voting)\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/21656715883745e0935ed6256a003ddd : SUCCESS in 43m 08s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/3e4e721affb34cea948bd562a84363b2 : SUCCESS in 33m 40s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/2245e382a0274b138b5afb58cd37a616 : SUCCESS in 58m 42s","accounts_in_message":[],"_revision_number":1},{"id":"9ea6d7445fc4ff6de89a90f8e604196380914476","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-18 07:50:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"932fdd28cde528907ded7f13c8c1b74dd480e56b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-18 07:51:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/724d9e32f1ee4f6c8279b560713ab0d7\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05014790c9a14786b69a9050b1d238b7 : SUCCESS in 35s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e71c19732ebeade352560b9c232e4a49388bac5a","revisions":{"e71c19732ebeade352560b9c232e4a49388bac5a":{"kind":"REWORK","_number":1,"created":"2024-11-11 05:40:26.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/64/934564/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/64/934564/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/64/934564/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/64/934564/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/64/934564/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/64/934564/1"}}},"commit":{"parents":[{"commit":"1980cc84d22f91e7bdd90f878822419bc4e1adf2","subject":"Merge \"Support file watcher to trigger GMR report\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/1980cc84d22f91e7bdd90f878822419bc4e1adf2"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-11-11 05:38:03.000000000","tz":-480},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-11-11 05:38:03.000000000","tz":-480},"subject":"Remove default override for config options policy_file","message":"Remove default override for config options policy_file\n\nolso.policy 4.5.0[1] changed the config options policy_file\ndefault value to \u0027policy.yaml\u0027, which means it is changed\nfor all the OpenStack services and they do not need to\noverride the default anymore.\n\nNOTE: There is no change in behaviour here, oslo.policy provides\nthe same configuration that services have overridden till now.\n\n[1] https://review.opendev.org/c/openstack/releases/+/934012\n[2] https://review.opendev.org/c/openstack/requirements/+/934295\n\nChange-Id: Iba548429fefef53d043d42de4d12554d77e4591a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/e71c19732ebeade352560b9c232e4a49388bac5a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/e71c19732ebeade352560b9c232e4a49388bac5a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
