)]}'
{"id":"openstack%2Faodh~935238","triplet_id":"openstack%2Faodh~master~I7a1dcb82c2f3ef848f69af6631846c617441458c","project":"openstack/aodh","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-11-15 15:11:16.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"I7a1dcb82c2f3ef848f69af6631846c617441458c","subject":"Replace deprecated configure_auth_token_middleware","status":"MERGED","created":"2024-11-15 01:59:24.000000000","updated":"2024-11-15 16:23:07.000000000","submitted":"2024-11-15 16:19:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"935238","meta_rev_id":"6c047def95c83a16c523067fbc87797b083887a2","_number":935238,"virtual_id_number":935238,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-15 16:19:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2024-11-15 15:11:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2024-11-15 15:11:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-15 03:26:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4850026b4dafe65d0b6ab34bd98d62afa274ce9d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-15 01:59:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7e7a8a89898be1f6614e6806dc1e2245f6399385","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-15 02:04:09.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"8fd33f5907579a2edbb1b780fb7fc2e7eb381cb9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-15 03:26:10.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/28c325706a744ef9bef58d7dc4f06bc4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d62578c4dd0f43cc9d60d37418c371e4 : SUCCESS in 3m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/18ec5fcb1786433984a14476c95da267 : SUCCESS in 5m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/207987b75d8344e8906f5c3c84530415 : SUCCESS in 6m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c31b621e07e74ecca73a19d26114f795 : SUCCESS in 7m 32s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/67cc0365073a46648f9b4c04d7e35a1f : FAILURE in 41m 54s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/e46ad5448a914d448367def87c9810bc : FAILURE in 17m 55s (non-voting)\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/61034c56e2ff42199915c32b4b62b0cb : FAILURE in 16m 56s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/5dce8f32071140de995e18c9aaaffcfd : FAILURE in 17m 42s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/bc1f60ae34e045ef9e6c92599c6abb75 : FAILURE in 34m 17s","accounts_in_message":[],"_revision_number":2},{"id":"f7067d113afd0046a6113a0cc1838a8c7ccbece0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-15 08:01:43.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"fee432ec4bd6c005644672fb2c51c6bf6319c82f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-15 09:04:03.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/03a53f1e68e14505941a03db2dc1b028\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/363a941a9ca64290aaab02256bdd6da2 : SUCCESS in 3m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/eb9d06dff9234d398e4262a055bbb920 : SUCCESS in 4m 59s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6edbb886a05a4a09bcf97094e5ffa851 : SUCCESS in 5m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ffcd69815fd94a6badef4cd3230ee556 : SUCCESS in 4m 27s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/3553528320c54e97bf5a91938ad671e7 : SUCCESS in 42m 06s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/b0093e3ae6b044f69438c423f96ae76d : SUCCESS in 34m 21s (non-voting)\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/1c6f734e985448cab5313c3a7163575e : SUCCESS in 28m 02s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/50e18067c94a4bd19786da31fe22e8a1 : SUCCESS in 41m 53s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/55b1fbf81a534d7c964ab9d273d85a34 : SUCCESS in 57m 13s","accounts_in_message":[],"_revision_number":3},{"id":"3333a2dfec9a1b07ce2306bf183a5a22a8b6a1f4","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-15 15:11:16.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"690097632cbccaecded38d3572e48ba835d57f1a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-15 15:11:59.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"dda5125ec734475881154960fc1b7b6f4fd5cbbf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-15 16:19:38.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5b4f227376ab463e8374d3f3a2ef3255\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2bd8f492536340ab9f3b14b2551c898f : SUCCESS in 2m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b4eda3621a6b4b9eae41062afef3ca41 : SUCCESS in 4m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f14aedc08584432899582c4a4ff81e5f : SUCCESS in 5m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6776297b31ad4dcfb72cfaa779adb09b : SUCCESS in 5m 33s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/7e5d3f79a612405a9feb772a5537332b : SUCCESS in 28m 30s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/35b4edd6e5af4ed886e1b468c44cafff : SUCCESS in 30m 05s (non-voting)\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/eea2f16131504f72895bac7ed3ee1883 : SUCCESS in 46m 55s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/9d02e2fdc078438da02825523f6128df : SUCCESS in 51m 49s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/0c96c86b885d40c9a678db982876c765 : SUCCESS in 49m 40s","accounts_in_message":[],"_revision_number":3},{"id":"4889ea0e39faaa2d200c6fd40d5d6eba7c27e7cd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-15 16:19:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"6c047def95c83a16c523067fbc87797b083887a2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-15 16:23:07.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ffe2c26363314317bb482b002e980bb9\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a6ee056c7c643ee89974e1ad0f24719 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"eaf0a39bea8b650d2339fe95857c75911bf06c40","revisions":{"32972c8c6e32e7aae8437efc986e1ebad426eac6":{"kind":"REWORK","_number":1,"created":"2024-11-15 01:59:24.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/38/935238/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/38/935238/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/38/935238/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/38/935238/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/38/935238/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/38/935238/1"}}},"commit":{"parents":[{"commit":"1980cc84d22f91e7bdd90f878822419bc4e1adf2","subject":"Merge \"Support file watcher to trigger GMR report\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/1980cc84d22f91e7bdd90f878822419bc4e1adf2"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-15 01:59:06.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-15 01:59:13.000000000","tz":540},"subject":"Replace deprecated configure_auth_token_middleware","message":"Replace deprecated configure_auth_token_middleware\n\nIt was deprecated some years ago by [1].\n\n[1] 8ab64b3236c93c8449edf80165017898b83f1269\n\nChange-Id: I7a1dcb82c2f3ef848f69af6631846c617441458c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/32972c8c6e32e7aae8437efc986e1ebad426eac6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/32972c8c6e32e7aae8437efc986e1ebad426eac6"}]},"branch":"refs/heads/master"},"42fd91dc6578a709e585fe10fcc880597c161446":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2024-11-15 02:04:09.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/38/935238/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/38/935238/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/38/935238/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/38/935238/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/38/935238/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/38/935238/2"}}},"commit":{"parents":[{"commit":"1980cc84d22f91e7bdd90f878822419bc4e1adf2","subject":"Merge \"Support file watcher to trigger GMR report\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/1980cc84d22f91e7bdd90f878822419bc4e1adf2"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-15 01:59:06.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-15 02:04:09.000000000","tz":0},"subject":"Replace deprecated configure_auth_token_middleware","message":"Replace deprecated configure_auth_token_middleware\n\nIt was deprecated some years ago by [1].\n\n[1] https://review.opendev.org/628651\n\nChange-Id: I7a1dcb82c2f3ef848f69af6631846c617441458c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/42fd91dc6578a709e585fe10fcc880597c161446"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/42fd91dc6578a709e585fe10fcc880597c161446"}]},"branch":"refs/heads/master","description":"Edit commit message"},"eaf0a39bea8b650d2339fe95857c75911bf06c40":{"kind":"REWORK","_number":3,"created":"2024-11-15 08:01:43.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/38/935238/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/38/935238/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/38/935238/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/38/935238/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/38/935238/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/38/935238/3"}}},"commit":{"parents":[{"commit":"1980cc84d22f91e7bdd90f878822419bc4e1adf2","subject":"Merge \"Support file watcher to trigger GMR report\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/1980cc84d22f91e7bdd90f878822419bc4e1adf2"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-15 01:59:06.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-15 08:01:30.000000000","tz":540},"subject":"Replace deprecated configure_auth_token_middleware","message":"Replace deprecated configure_auth_token_middleware\n\nIt was deprecated some years ago by [1].\n\n[1] 8ab64b3236c93c8449edf80165017898b83f1269\n\nChange-Id: I7a1dcb82c2f3ef848f69af6631846c617441458c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/eaf0a39bea8b650d2339fe95857c75911bf06c40"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/eaf0a39bea8b650d2339fe95857c75911bf06c40"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
