)]}'
{"id":"openstack%2Faodh~935479","triplet_id":"openstack%2Faodh~stable%2F2024.2~Iec6ab6b4fcdb60caa67197acb5ada0678c0309ce","project":"openstack/aodh","branch":"stable/2024.2","topic":"global-venv-stable/2024.2","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-04-20 13:26:09.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"Iec6ab6b4fcdb60caa67197acb5ada0678c0309ce","subject":"Revert \"Disable GLOBAL_VENV\"","status":"ABANDONED","created":"2024-11-18 06:01:08.000000000","updated":"2026-04-20 13:26:12.000000000","total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":932526,"cherry_pick_of_patch_set":2,"meta_rev_id":"e90a77787fd928f2e25a2617edcbb0a47ba55ab8","_number":935479,"virtual_id_number":935479,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-04-29 07:28:07.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"all":[{"value":2,"date":"2025-01-24 13:42:01.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":2,"date":"2024-11-18 06:41:35.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"all":[{"value":1,"date":"2025-01-24 13:42:01.000000000","_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"}],"REVIEWER":[{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-18 06:39:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-18 06:41:35.000000000","updated_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"reviewer":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"state":"REVIEWER"},{"updated":"2025-01-24 13:42:01.000000000","updated_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"},{"updated":"2026-03-27 08:46:20.000000000","updated_by":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"reviewer":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"state":"CC"}],"messages":[{"id":"34ebc728719aa7ab10e74bbd342a6815fbb67bd2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-18 06:01:08.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"de49f1b6e1f719510ce10a515789726a07345a88","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-18 06:39:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/15122ee0f4474b4492865050b2014321\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/becb34c93a9e479290bd533588bfd015 : SUCCESS in 2m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5f05327fefc544b096f045fd192d300f : SUCCESS in 3m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3540738a912c4f4bb53dae41bbe1947f : SUCCESS in 3m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/07062230d1f74687b9bf03ca1112fc35 : SUCCESS in 3m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57ce5528d8d84d1db600c611d6c98945 : SUCCESS in 7m 19s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/38bf33314422482e938acd419f53290f : SUCCESS in 36m 51s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/387de73448114fd4a5bdcd638f92fab6 : SUCCESS in 20m 13s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/30bf6b9e00f8412fa91381045dd5a895 : SUCCESS in 26m 13s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/63dd7142f8e740fd86610b4b7d2e0928 : SUCCESS in 26m 04s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/06cea21e012d45c99611743491d487db : SUCCESS in 31m 56s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/c009620a7c204833aead8146327bee4a : SUCCESS in 34m 30s","accounts_in_message":[],"_revision_number":1},{"id":"7b6ca261c46a7030d66b67828f09f72913233c65","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2024-11-18 06:41:35.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"bcdbcf4446a5ae71d8eacc2bc764e26c280abca2","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2025-01-24 13:42:01.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"755640e14e533396752fa7afd1152c2b587330ce","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-24 13:42:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5197c1bc2bcd8a0eedb093a2151ba430c79e632f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-24 15:31:48.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f9a65353ecdd40a1bfbfe8373de87a0e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aaa72068e9a64b08bb26215127356d52 : SUCCESS in 4m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4541afaf1b3a4c18aad799187091511e : SUCCESS in 4m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1d075f64aa4b41b880969ebb7384c061 : SUCCESS in 4m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fda7f24c572044c386a886720fb4cbb2 : SUCCESS in 6m 53s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/1b4a3522ad6e4a3cb0233770cde4dc2c : FAILURE in 22m 58s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/9df075d55edf4542a4fba40b355fc200 : FAILURE in 30m 26s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/4147d4447fe7468ea57935e9e1ec4814 : SUCCESS in 41m 40s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/4f14de7d7e96465ba7e381e25ede2047 : SUCCESS in 1h 40m 52s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/8996a0faab534090b93e97884f78728d : SUCCESS in 1h 32m 29s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/a8e2845653144aaf95963d5f7496c410 : SUCCESS in 45m 05s","accounts_in_message":[],"_revision_number":1},{"id":"213c4597825f7abcf6b75af932b8ef61ba2b1d00","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2025-04-29 06:46:36.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e4ebcd46d4f20a297fcbbf79818704a5572b3dec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-29 07:28:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d62cf6ce6adc47959b53dcff0b79efc6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a5701a5edca40f29add3a32371a6045 : SUCCESS in 3m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/93dd1affbe8d4860b62d7b5fdc88122e : SUCCESS in 4m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5b0f86a92bc643d3b8c211ff796e3a53 : SUCCESS in 4m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/220425baaae54bcaaf7de2a1ebea4143 : SUCCESS in 2m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ab9e2065ff34a18bb6026fd3e0787dc : SUCCESS in 5m 35s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/0b50b15aa4c543cb919418cad28e9b92 : FAILURE in 7m 25s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/17779386985e41c89ce60de7cff29230 : FAILURE in 10m 48s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/d1ad309d41164f1ebc23ba5099ae8634 : SUCCESS in 29m 03s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/4929ac06939b415ab0e04f4db1f0cf6a : SUCCESS in 20m 06s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/e7fa8090dec74ccc88a11d5000dcdfcc : FAILURE in 36m 29s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/d2c35d4cdd914195b875bde411f4fcd4 : FAILURE in 28m 33s","accounts_in_message":[],"_revision_number":1},{"id":"c3945130df3d1f8f039c0040c2557cf74667eb24","author":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"date":"2026-03-27 08:46:20.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"72cc880b74b4ae8cd33cdd779846d4bc2bbb6aff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-27 09:41:39.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/dbc4ab75244949c1a4410c4fb3fd44d0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b76358af3cce42a1bf4c100e6682f52b : SUCCESS in 4m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d45f0c69466d43fa9b790211a93a313d : FAILURE in 3m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0b5a959c32834c45a154b110378da6f8 : FAILURE in 3m 11s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6c83e31ba84743e7bca3423c564cae09 : FAILURE in 5m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d327568a69747919ae5965683d60f71 : FAILURE in 2m 52s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/f8e0ded8b1e347b0a646b38b30c55727 : FAILURE in 9m 00s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/eb13d0d0c96a49db98586f58f462dd55 : FAILURE in 15m 22s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/93818e1780dc45a89d88ede956afa0de : SUCCESS in 26m 57s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/c8685b8da86a4416a226db3ad5ef16cd : SUCCESS in 49m 37s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/ffcb90263f034c4f943edafd25e8910a : FAILURE in 13m 01s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/11b6847c8c234898a426d7568bd7d55a : FAILURE in 21m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b9130d319d18bc1ca8f9c0de282a468339427da1","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-04-20 13:26:09.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e90a77787fd928f2e25a2617edcbb0a47ba55ab8","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-04-20 13:26:12.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f99cbc99c2cbf725bee205af2328a0dc988fb1e4","revisions":{"f99cbc99c2cbf725bee205af2328a0dc988fb1e4":{"kind":"REWORK","_number":1,"created":"2024-11-18 06:01:08.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/79/935479/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/79/935479/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/79/935479/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/79/935479/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/79/935479/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/79/935479/1"}}},"commit":{"parents":[{"commit":"faef3b0c7a84704c10d99754179989e656db5c53","subject":"Update TOX_CONSTRAINTS_FILE for stable/2024.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/faef3b0c7a84704c10d99754179989e656db5c53"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-16 14:40:16.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-18 06:01:08.000000000","tz":0},"subject":"Revert \"Disable GLOBAL_VENV\"","message":"Revert \"Disable GLOBAL_VENV\"\n\nThis reverts commit 8de7585f374097bcb4910bfbed34b6c721490db9.\n\nReason for revert:\nCurrent devstack disables GLOBAL_VENV automatically in CentOS or Rocky\nso we don\u0027t have to override the option. GLOBAL_VENV\u003dTrue is required\nto migrate the jobs to Ubuntu 24.04 (Noble) .\n\nThis also changes the default API deployment method to uwsgi because\nuwsgi is now globally used in CI and only works for global venv\nmethod.\n\nConflicts:\n\t.zuul.yaml\n\nChange-Id: Iec6ab6b4fcdb60caa67197acb5ada0678c0309ce\n(cherry picked from commit 1c4761eccae61acad5536b41e3b3835e93caf64d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/f99cbc99c2cbf725bee205af2328a0dc988fb1e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/f99cbc99c2cbf725bee205af2328a0dc988fb1e4"}]},"branch":"refs/heads/stable/2024.2"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
