)]}'
{"id":"openstack%2Faodh~936060","triplet_id":"openstack%2Faodh~master~I1a876d0e1273fdc1410532864230f801cff3ae1f","project":"openstack/aodh","branch":"master","topic":"normalize-wsgi","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-11-24 18:31:50.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I1a876d0e1273fdc1410532864230f801cff3ae1f","subject":"devstack: Remove mod_wsgi support","status":"ABANDONED","created":"2024-11-22 19:44:35.000000000","updated":"2024-11-24 18:31:50.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"6d6608aa11ead4bb06bae70604193bd69c4d2050","_number":936060,"virtual_id_number":936060,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-11-23 08:57:10.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-22 22:12:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"26fbd1f26933723ccdc61c80a052ca7f79c47c8f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-22 19:44:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d0a92d3c60b67e12d8d7d2ef30f591257f02b1cd","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-22 19:44:52.000000000","message":"Topic set to normalize-wsgi","accounts_in_message":[],"_revision_number":1},{"id":"c9eb31d9f4c1b6aee72c3b0a524d40926b7f128a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-22 19:47:16.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"39adc2073d5ce9dbe57168be29f647fd2636d341","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-22 20:04:41.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"50d75a939f47edb7fec7b5b713c1ae685163cfa6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-22 20:08:11.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"facf19651b630d86662a3ff7a490ab535fab5614","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-22 20:12:26.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"f29d8d52e8847748208efa415bb4ff5384474d13","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-22 20:19:34.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"2fd550d21cbec85582df5f8676de5fa1366480d9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-22 20:20:47.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"a049804c9ce7e906e3c1326598235c5deb87077d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-22 20:58:12.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"f1b605dd624a0310423b86ef135c9960a2c3b59f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-22 21:01:12.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"f0fc990e8231d7ea8e81489991a0037bd789a715","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-22 21:24:05.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"bd938ac7122b2831f731c506809f055ca570bccb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-22 21:24:25.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"e00a736d72f29758be7b5e3408a5aa91ee8c5497","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-22 22:12:14.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/867f449b92954504a805c386ad8e1610\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/93cd47ad98994423bcb38273762c6f64 : SUCCESS in 3m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d086fd34506b444fae0dd73dbf4d78f8 : SUCCESS in 5m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3853af98745948299226fe41c17cc650 : SUCCESS in 5m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92d4510604044d5dba2f8ea361f5ec3e : SUCCESS in 6m 04s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/e1e3f0c381994646a0544c17b3bbf1ee : FAILURE in 18m 16s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/4bba67f14ffc4db8968d3963cff33f4e : FAILURE in 14m 30s (non-voting)\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/45373deaa6f54abe9eca433182cf8d45 : FAILURE in 17m 16s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/14ab371e66d04e368ec1c4207bf68ceb : FAILURE in 24m 21s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/316eaf0eda6e49358698e9d372406692 : FAILURE in 45m 52s","accounts_in_message":[],"_revision_number":11},{"id":"183552d491e56a892b1bac15da1dbdd6b545f101","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-23 04:32:02.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":12},{"id":"15f4142f3004a6e78083d0f4cebd0ebba614ba8a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-23 05:15:03.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2a7bb6ddbcff496aabbe370630f416c7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0f9875cf8ec49248befd4c77c634281 : SUCCESS in 2m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/269c6ee5ba6946d39f4a07d227277517 : SUCCESS in 5m 04s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b3f08e92ac09408ea079ef189018deaf : SUCCESS in 5m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e5b537b5a86d425fa06f7d44bd68d216 : SUCCESS in 6m 36s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/2767f6571577488b9d0cb0ee0ef76fe7 : FAILURE in 33m 18s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/ca8de8b699ad49c9943390a34388b5b0 : FAILURE in 29m 52s (non-voting)\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/68eea94cdd104ac0a1cfbf99f6a581ce : FAILURE in 30m 27s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/11c41e18a9534ed98874bc3e3bf9bb12 : FAILURE in 22m 24s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/74f45daf95a94947a934507877506517 : FAILURE in 24m 59s","accounts_in_message":[],"_revision_number":12},{"id":"ece616c6459033a45767171b19f91ad7c864dcad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-23 06:30:30.000000000","message":"Uploaded patch set 13.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":13},{"id":"a351c423ef05ccfe479b63c74bda11b95ced8ed2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-23 06:55:01.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/39c716734f1446129adb50fb66ec0981\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1db42f0c9c944e199defc96ae06e06fc : SUCCESS in 3m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b1511a49fe5f42a08921b4fbe0743a21 : SUCCESS in 4m 03s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/93c32aff565841aca96325af1dd081a6 : SUCCESS in 4m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ff256c65a8645f7aab78a99d937f5e1 : SUCCESS in 5m 43s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/63d4b19290d74b169f394e4f9f87fd3e : FAILURE in 16m 42s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/55e6e808b8af441ea67bb0e77a440427 : FAILURE in 14m 40s (non-voting)\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/b5a8a7251650444db5b2a1eccc201d66 : FAILURE in 15m 49s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/acfbcc45c9bb4251ab0f3a7b2ce4ad49 : FAILURE in 20m 40s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/59fc087be1954cae850612efef634387 : FAILURE in 23m 33s","accounts_in_message":[],"_revision_number":13},{"id":"29fbdd7bfc9f819d3bcad13bce0c9509c5a700f9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-23 06:58:25.000000000","message":"Uploaded patch set 14.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":14},{"id":"1cb2d43a503bbef17d20d9d2da60ae6f69df8f0b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-23 07:43:58.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c97e5180bfa64805b225660f0f26d483\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b763e71e925b4286b2958a8d4c857020 : SUCCESS in 5m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1c8370da5d3f46cbac0196472d32b0b5 : SUCCESS in 5m 49s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/99f5768142e64993aa752199d7d8ca5c : SUCCESS in 4m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e3b74bf811204e958117c1dbc71be7b4 : SUCCESS in 5m 09s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/c0d7c9c54465412f91cc5cd7b4fc1d89 : FAILURE in 32m 12s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/554c0386b91b40758b84eb97fb6df8dd : FAILURE in 15m 05s (non-voting)\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/a40be27740d14a41abb28150a6325b58 : FAILURE in 33m 23s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/a138f7c741ac428fbc227886527786f3 : FAILURE in 40m 03s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/87085ff264654aff87f2b0207c887d42 : FAILURE in 21m 36s","accounts_in_message":[],"_revision_number":14},{"id":"203cf83f89776d04b31b9681f7e781574ceaa9a1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-23 08:05:54.000000000","message":"Uploaded patch set 15.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":15},{"id":"8a1e919cf37e6833d39e5739d9538a95a7b28310","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-23 08:08:12.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"4cb8aff75433c1f50678b10f97c9f9aa08ec73d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-23 08:57:10.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0fcd2330e5a9417d98815d95e9d43e23\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d6f1066f05a4b53928f341150c8c819 : SUCCESS in 4m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2d3c59db97f8474593c228c7a532d4ca : SUCCESS in 3m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f380949939e04560bbc2785fc710e2ee : SUCCESS in 5m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c6214e7d29b46ee94d758eb7e004e13 : SUCCESS in 4m 26s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/315cf50acef242a7b585e5c4f806f180 : FAILURE in 38m 55s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/3e8f5c626c574d6aae2573fc049afcaa : FAILURE in 17m 45s (non-voting)\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/e5bbb62701784dc9975a47f7f440d748 : FAILURE in 29m 30s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/dcf8b0a547744fdf8e3064b65fc522ce : FAILURE in 19m 28s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/134c4cdd38be48efa4c936567d42a087 : FAILURE in 42m 54s","accounts_in_message":[],"_revision_number":16},{"id":"6d6608aa11ead4bb06bae70604193bd69c4d2050","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-24 18:31:50.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":16}],"current_revision_number":16,"current_revision":"dd05966c6dc65fbd949282989df24a7feac0a382","revisions":{"8290eadabb94f70fd18eee25ebb2e8cce0d2cd4f":{"kind":"REWORK","_number":1,"created":"2024-11-22 19:44:35.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/60/936060/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/60/936060/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/60/936060/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/60/936060/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/60/936060/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/60/936060/1"}}},"commit":{"parents":[{"commit":"10cde330b5ab0340dbef17bba07df6c4c429c1a8","subject":"reno: Update master for unmaintained/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/10cde330b5ab0340dbef17bba07df6c4c429c1a8"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-22 19:43:30.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-22 19:44:10.000000000","tz":540},"subject":"devstack: Remove mod_wsgi support","message":"devstack: Remove mod_wsgi support\n\nThe same is being removed from devstack (for example:\nhttps://review.opendev.org/c/openstack/devstack/+/932195 )\n\nNote that this also removes the remaining logic to support standalone\naodh-api because the aodh-api console script does not exist.\n\nChange-Id: I1a876d0e1273fdc1410532864230f801cff3ae1f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/8290eadabb94f70fd18eee25ebb2e8cce0d2cd4f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/8290eadabb94f70fd18eee25ebb2e8cce0d2cd4f"}]},"branch":"refs/heads/master"},"aad0a8f16957bce523eece37f661258e3433464f":{"kind":"REWORK","_number":2,"created":"2024-11-22 19:47:16.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/60/936060/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/60/936060/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/60/936060/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/60/936060/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/60/936060/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/60/936060/2"}}},"commit":{"parents":[{"commit":"10cde330b5ab0340dbef17bba07df6c4c429c1a8","subject":"reno: Update master for unmaintained/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/10cde330b5ab0340dbef17bba07df6c4c429c1a8"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-22 19:43:30.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-22 19:47:04.000000000","tz":540},"subject":"devstack: Remove mod_wsgi support","message":"devstack: Remove mod_wsgi support\n\nThe same is being removed from devstack (for example:\nhttps://review.opendev.org/c/openstack/devstack/+/932195 )\n\nNote that this also removes the remaining logic to support standalone\naodh-api because the aodh-api console script does not exist.\n\nChange-Id: I1a876d0e1273fdc1410532864230f801cff3ae1f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/aad0a8f16957bce523eece37f661258e3433464f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/aad0a8f16957bce523eece37f661258e3433464f"}]},"branch":"refs/heads/master"},"b2a1fed8f98ccaca25fecd0b6b5730aff2788f00":{"kind":"REWORK","_number":3,"created":"2024-11-22 20:04:41.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/60/936060/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/60/936060/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/60/936060/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/60/936060/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/60/936060/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/60/936060/3"}}},"commit":{"parents":[{"commit":"10cde330b5ab0340dbef17bba07df6c4c429c1a8","subject":"reno: Update master for unmaintained/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/10cde330b5ab0340dbef17bba07df6c4c429c1a8"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-22 19:43:30.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-22 20:04:23.000000000","tz":540},"subject":"devstack: Remove mod_wsgi support","message":"devstack: Remove mod_wsgi support\n\nThe same is being removed from devstack (for example:\nhttps://review.opendev.org/c/openstack/devstack/+/932195 )\n\nNote that this also removes the remaining logic to support standalone\naodh-api because the aodh-api console script does not exist.\n\nChange-Id: I1a876d0e1273fdc1410532864230f801cff3ae1f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/b2a1fed8f98ccaca25fecd0b6b5730aff2788f00"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/b2a1fed8f98ccaca25fecd0b6b5730aff2788f00"}]},"branch":"refs/heads/master"},"0be76e00630a9c1ded0e181f5ff46321b8cbb37b":{"kind":"REWORK","_number":4,"created":"2024-11-22 20:08:11.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/60/936060/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/60/936060/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/60/936060/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/60/936060/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/60/936060/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/60/936060/4"}}},"commit":{"parents":[{"commit":"10cde330b5ab0340dbef17bba07df6c4c429c1a8","subject":"reno: Update master for unmaintained/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/10cde330b5ab0340dbef17bba07df6c4c429c1a8"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-22 19:43:30.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-22 20:08:01.000000000","tz":540},"subject":"devstack: Remove mod_wsgi support","message":"devstack: Remove mod_wsgi support\n\nThe same is being removed from devstack (for example:\nhttps://review.opendev.org/c/openstack/devstack/+/932195 )\n\nNote that this also removes the remaining logic to support standalone\naodh-api because the aodh-api console script does not exist.\n\nChange-Id: I1a876d0e1273fdc1410532864230f801cff3ae1f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/0be76e00630a9c1ded0e181f5ff46321b8cbb37b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/0be76e00630a9c1ded0e181f5ff46321b8cbb37b"}]},"branch":"refs/heads/master"},"6b86b146fbb31214ff1f83cb82c9d20a9c0afe10":{"kind":"REWORK","_number":5,"created":"2024-11-22 20:12:26.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/60/936060/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/60/936060/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/60/936060/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/60/936060/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/60/936060/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/60/936060/5"}}},"commit":{"parents":[{"commit":"10cde330b5ab0340dbef17bba07df6c4c429c1a8","subject":"reno: Update master for unmaintained/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/10cde330b5ab0340dbef17bba07df6c4c429c1a8"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-22 19:43:30.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-22 20:12:26.000000000","tz":0},"subject":"devstack: Remove mod_wsgi support","message":"devstack: Remove mod_wsgi support\n\nThe same is being removed from devstack (for example:\nhttps://review.opendev.org/c/openstack/devstack/+/932195 )\n\nNote that this also removes the remaining logic to support standalone\naodh-api because the aodh-api console script does not exist.\n\nChange-Id: I1a876d0e1273fdc1410532864230f801cff3ae1f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/6b86b146fbb31214ff1f83cb82c9d20a9c0afe10"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/6b86b146fbb31214ff1f83cb82c9d20a9c0afe10"}]},"branch":"refs/heads/master"},"2900cb903e595e6875dcb39a96bac2e851b064ca":{"kind":"REWORK","_number":6,"created":"2024-11-22 20:19:34.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/60/936060/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/60/936060/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/60/936060/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/60/936060/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/60/936060/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/60/936060/6"}}},"commit":{"parents":[{"commit":"10cde330b5ab0340dbef17bba07df6c4c429c1a8","subject":"reno: Update master for unmaintained/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/10cde330b5ab0340dbef17bba07df6c4c429c1a8"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-22 19:43:30.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-22 20:19:23.000000000","tz":540},"subject":"devstack: Remove mod_wsgi support","message":"devstack: Remove mod_wsgi support\n\nThe same is being removed from devstack (for example:\nhttps://review.opendev.org/c/openstack/devstack/+/932195 )\n\nNote that this also removes the remaining logic to support standalone\naodh-api because the aodh-api console script does not exist.\n\nChange-Id: I1a876d0e1273fdc1410532864230f801cff3ae1f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/2900cb903e595e6875dcb39a96bac2e851b064ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/2900cb903e595e6875dcb39a96bac2e851b064ca"}]},"branch":"refs/heads/master"},"5bc4273a00166bd7523bb56cd2539b578894cbd6":{"kind":"REWORK","_number":7,"created":"2024-11-22 20:20:47.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/60/936060/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/60/936060/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/60/936060/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/60/936060/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/60/936060/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/60/936060/7"}}},"commit":{"parents":[{"commit":"10cde330b5ab0340dbef17bba07df6c4c429c1a8","subject":"reno: Update master for unmaintained/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/10cde330b5ab0340dbef17bba07df6c4c429c1a8"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-22 19:43:30.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-22 20:20:18.000000000","tz":540},"subject":"devstack: Remove mod_wsgi support","message":"devstack: Remove mod_wsgi support\n\nThe same is being removed from devstack (for example:\nhttps://review.opendev.org/c/openstack/devstack/+/932195 )\n\nNote that this also removes the remaining logic to support standalone\naodh-api because the aodh-api console script does not exist.\n\nChange-Id: I1a876d0e1273fdc1410532864230f801cff3ae1f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/5bc4273a00166bd7523bb56cd2539b578894cbd6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/5bc4273a00166bd7523bb56cd2539b578894cbd6"}]},"branch":"refs/heads/master"},"c7e53d8fefff51eda82418672bcdf055e904025d":{"kind":"REWORK","_number":8,"created":"2024-11-22 20:58:12.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/60/936060/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/60/936060/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/60/936060/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/60/936060/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/60/936060/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/60/936060/8"}}},"commit":{"parents":[{"commit":"10cde330b5ab0340dbef17bba07df6c4c429c1a8","subject":"reno: Update master for unmaintained/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/10cde330b5ab0340dbef17bba07df6c4c429c1a8"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-22 19:43:30.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-22 20:58:01.000000000","tz":540},"subject":"devstack: Remove mod_wsgi support","message":"devstack: Remove mod_wsgi support\n\nThe same is being removed from devstack (for example:\nhttps://review.opendev.org/c/openstack/devstack/+/932195 )\n\nNote that this also removes the remaining logic to support standalone\naodh-api because the aodh-api console script does not exist.\n\nChange-Id: I1a876d0e1273fdc1410532864230f801cff3ae1f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/c7e53d8fefff51eda82418672bcdf055e904025d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/c7e53d8fefff51eda82418672bcdf055e904025d"}]},"branch":"refs/heads/master"},"af2ca87c92b925830a4d92561bb710b8d2de62f5":{"kind":"REWORK","_number":9,"created":"2024-11-22 21:01:12.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/60/936060/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/60/936060/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/60/936060/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/60/936060/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/60/936060/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/60/936060/9"}}},"commit":{"parents":[{"commit":"10cde330b5ab0340dbef17bba07df6c4c429c1a8","subject":"reno: Update master for unmaintained/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/10cde330b5ab0340dbef17bba07df6c4c429c1a8"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-22 19:43:30.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-22 21:00:53.000000000","tz":540},"subject":"devstack: Remove mod_wsgi support","message":"devstack: Remove mod_wsgi support\n\nThe same is being removed from devstack (for example:\nhttps://review.opendev.org/c/openstack/devstack/+/932195 )\n\nNote that this also removes the remaining logic to support standalone\naodh-api because the aodh-api console script does not exist.\n\nChange-Id: I1a876d0e1273fdc1410532864230f801cff3ae1f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/af2ca87c92b925830a4d92561bb710b8d2de62f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/af2ca87c92b925830a4d92561bb710b8d2de62f5"}]},"branch":"refs/heads/master"},"8358f4a0557ec00a2afa12f8fb7d9cc3d9c49831":{"kind":"REWORK","_number":10,"created":"2024-11-22 21:24:05.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/60/936060/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/60/936060/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/60/936060/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/60/936060/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/60/936060/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/60/936060/10"}}},"commit":{"parents":[{"commit":"10cde330b5ab0340dbef17bba07df6c4c429c1a8","subject":"reno: Update master for unmaintained/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/10cde330b5ab0340dbef17bba07df6c4c429c1a8"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-22 19:43:30.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-22 21:23:55.000000000","tz":540},"subject":"devstack: Remove mod_wsgi support","message":"devstack: Remove mod_wsgi support\n\nThe same is being removed from devstack (for example:\nhttps://review.opendev.org/c/openstack/devstack/+/932195 )\n\nNote that this also removes the remaining logic to support standalone\naodh-api because the aodh-api console script does not exist.\n\nChange-Id: I1a876d0e1273fdc1410532864230f801cff3ae1f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/8358f4a0557ec00a2afa12f8fb7d9cc3d9c49831"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/8358f4a0557ec00a2afa12f8fb7d9cc3d9c49831"}]},"branch":"refs/heads/master"},"14c451d48777e832f268a3a353e1f6aaddcd22f7":{"kind":"REWORK","_number":11,"created":"2024-11-22 21:24:25.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/60/936060/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/60/936060/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/60/936060/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/60/936060/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/60/936060/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/60/936060/11"}}},"commit":{"parents":[{"commit":"10cde330b5ab0340dbef17bba07df6c4c429c1a8","subject":"reno: Update master for unmaintained/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/10cde330b5ab0340dbef17bba07df6c4c429c1a8"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-22 19:43:30.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-22 21:24:15.000000000","tz":540},"subject":"devstack: Remove mod_wsgi support","message":"devstack: Remove mod_wsgi support\n\nThe same is being removed from devstack (for example:\nhttps://review.opendev.org/c/openstack/devstack/+/932195 )\n\nNote that this also removes the remaining logic to support standalone\naodh-api because the aodh-api console script does not exist.\n\nChange-Id: I1a876d0e1273fdc1410532864230f801cff3ae1f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/14c451d48777e832f268a3a353e1f6aaddcd22f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/14c451d48777e832f268a3a353e1f6aaddcd22f7"}]},"branch":"refs/heads/master"},"c488afd4ce4b0cbb4e29a83342a10cebf609ab48":{"kind":"REWORK","_number":12,"created":"2024-11-23 04:32:02.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/60/936060/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/60/936060/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/60/936060/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/60/936060/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/60/936060/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/60/936060/12"}}},"commit":{"parents":[{"commit":"10cde330b5ab0340dbef17bba07df6c4c429c1a8","subject":"reno: Update master for unmaintained/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/10cde330b5ab0340dbef17bba07df6c4c429c1a8"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-22 19:43:30.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-23 04:31:51.000000000","tz":540},"subject":"devstack: Remove mod_wsgi support","message":"devstack: Remove mod_wsgi support\n\nThe same is being removed from devstack (for example:\nhttps://review.opendev.org/c/openstack/devstack/+/932195 )\n\nNote that this also removes the remaining logic to support standalone\naodh-api because the aodh-api console script does not exist.\n\nChange-Id: I1a876d0e1273fdc1410532864230f801cff3ae1f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/c488afd4ce4b0cbb4e29a83342a10cebf609ab48"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/c488afd4ce4b0cbb4e29a83342a10cebf609ab48"}]},"branch":"refs/heads/master"},"ecef7d8a88cfc7a7ba16a6a9b438e1e8e7959591":{"kind":"REWORK","_number":13,"created":"2024-11-23 06:30:30.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/60/936060/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/60/936060/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/60/936060/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/60/936060/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/60/936060/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/60/936060/13"}}},"commit":{"parents":[{"commit":"10cde330b5ab0340dbef17bba07df6c4c429c1a8","subject":"reno: Update master for unmaintained/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/10cde330b5ab0340dbef17bba07df6c4c429c1a8"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-22 19:43:30.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-23 06:30:20.000000000","tz":540},"subject":"devstack: Remove mod_wsgi support","message":"devstack: Remove mod_wsgi support\n\nThe same is being removed from devstack (for example:\nhttps://review.opendev.org/c/openstack/devstack/+/932195 )\n\nNote that this also removes the remaining logic to support standalone\naodh-api because the aodh-api console script does not exist.\n\nChange-Id: I1a876d0e1273fdc1410532864230f801cff3ae1f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/ecef7d8a88cfc7a7ba16a6a9b438e1e8e7959591"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/ecef7d8a88cfc7a7ba16a6a9b438e1e8e7959591"}]},"branch":"refs/heads/master"},"8fdd4c0d945c9dad48e2ee696c249265cbc88174":{"kind":"REWORK","_number":14,"created":"2024-11-23 06:58:25.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/60/936060/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/60/936060/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/60/936060/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/60/936060/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/60/936060/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/60/936060/14"}}},"commit":{"parents":[{"commit":"10cde330b5ab0340dbef17bba07df6c4c429c1a8","subject":"reno: Update master for unmaintained/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/10cde330b5ab0340dbef17bba07df6c4c429c1a8"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-22 19:43:30.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-23 06:58:15.000000000","tz":540},"subject":"devstack: Remove mod_wsgi support","message":"devstack: Remove mod_wsgi support\n\nThe same is being removed from devstack (for example:\nhttps://review.opendev.org/c/openstack/devstack/+/932195 )\n\nNote that this also removes the remaining logic to support standalone\naodh-api because the aodh-api console script does not exist.\n\nChange-Id: I1a876d0e1273fdc1410532864230f801cff3ae1f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/8fdd4c0d945c9dad48e2ee696c249265cbc88174"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/8fdd4c0d945c9dad48e2ee696c249265cbc88174"}]},"branch":"refs/heads/master"},"c881789186e46256c84621da1e9c9724c15ec8f3":{"kind":"REWORK","_number":15,"created":"2024-11-23 08:05:54.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/60/936060/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/60/936060/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/60/936060/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/60/936060/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/60/936060/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/60/936060/15"}}},"commit":{"parents":[{"commit":"10cde330b5ab0340dbef17bba07df6c4c429c1a8","subject":"reno: Update master for unmaintained/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/10cde330b5ab0340dbef17bba07df6c4c429c1a8"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-22 19:43:30.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-23 08:05:44.000000000","tz":540},"subject":"devstack: Remove mod_wsgi support","message":"devstack: Remove mod_wsgi support\n\nThe same is being removed from devstack (for example:\nhttps://review.opendev.org/c/openstack/devstack/+/932195 )\n\nNote that this also removes the remaining logic to support standalone\naodh-api because the aodh-api console script does not exist.\n\nChange-Id: I1a876d0e1273fdc1410532864230f801cff3ae1f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/c881789186e46256c84621da1e9c9724c15ec8f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/c881789186e46256c84621da1e9c9724c15ec8f3"}]},"branch":"refs/heads/master"},"dd05966c6dc65fbd949282989df24a7feac0a382":{"kind":"REWORK","_number":16,"created":"2024-11-23 08:08:12.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/60/936060/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/60/936060/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/60/936060/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/60/936060/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/60/936060/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/60/936060/16"}}},"commit":{"parents":[{"commit":"10cde330b5ab0340dbef17bba07df6c4c429c1a8","subject":"reno: Update master for unmaintained/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/10cde330b5ab0340dbef17bba07df6c4c429c1a8"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-22 19:43:30.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-23 08:08:00.000000000","tz":540},"subject":"devstack: Remove mod_wsgi support","message":"devstack: Remove mod_wsgi support\n\nThe same is being removed from devstack (for example:\nhttps://review.opendev.org/c/openstack/devstack/+/932195 )\n\nNote that this also removes the remaining logic to support standalone\naodh-api because the aodh-api console script does not exist.\n\nChange-Id: I1a876d0e1273fdc1410532864230f801cff3ae1f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/dd05966c6dc65fbd949282989df24a7feac0a382"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/dd05966c6dc65fbd949282989df24a7feac0a382"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
