)]}'
{"id":"openstack%2Faodh~936078","triplet_id":"openstack%2Faodh~master~I56de124a3a6d4729cfd12d79deab59e96fed9fdd","project":"openstack/aodh","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-02-11 05:26:53.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"I56de124a3a6d4729cfd12d79deab59e96fed9fdd","subject":"Drop remaining reference to \u0027test\u0027 extras","status":"MERGED","created":"2024-11-23 09:00:02.000000000","updated":"2025-02-11 06:42:12.000000000","submitted":"2025-02-11 06:41:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"936078","meta_rev_id":"69444d5ce30f1b5fac5d3a762373910582afb331","_number":936078,"virtual_id_number":936078,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-02-11 06:41:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2025-02-11 05:26:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2025-02-11 05:26:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-23 09:21:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5552c8b6ac063909d02fcf16782244717b2aa480","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-23 09:00:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a0be37cad5d9d2c703bbff2d5db87ca66575dbbd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-23 09:21:53.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/de6b550581744b3ca28cae0670b60f48\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/74106f99d6b84d83a754327a07e4d0cf : SUCCESS in 3m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/57f5d900efa048a78cb0ea3056d687a7 : SUCCESS in 4m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b45e07c60cd042c4ace88d3904159fb1 : SUCCESS in 5m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82376f0aa4b54007ace0bbc3302800c1 : SUCCESS in 5m 44s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/ca2c05cba3c04054a471430d80750f92 : FAILURE in 10m 38s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/23d52498fcbf45a083c3cb3c48e3daa3 : FAILURE in 11m 25s (non-voting)\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/ed7fde8523c24451bf410bf492d5b69c : FAILURE in 10m 33s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/0602716e157c4a9dbeda76a67d3aa114 : FAILURE in 18m 59s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/6072a2ce144a4594b2ba23493d3b9e60 : FAILURE in 17m 35s","accounts_in_message":[],"_revision_number":1},{"id":"3d2ae34ad6e54ad7dbc940e1640e0913857d964a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-23 09:35:10.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"17e000d0903fb78ba885d735b84e098bfa97e954","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-23 10:20:16.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bb547fb0f5d04ca48a73ecd5d9ad0acb\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/706e2899118b43e98d399b4a28c9d8f7 : SUCCESS in 2m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cb47cf90f9984347b4a6c805b767521f : SUCCESS in 4m 50s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f2b5e1c3f4964dfe9c34622115a2907a : SUCCESS in 5m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f349a0ca4854b5d832de74c60bbc91b : SUCCESS in 4m 10s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/382040ff0361462b8eba46b23ae4c51b : SUCCESS in 35m 47s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/bc5fede382f044ff8370921adaeb9515 : SUCCESS in 37m 05s (non-voting)\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/7f9417fb5ff548a4b41f895640a0951c : SUCCESS in 27m 33s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/6b221fb6a0014851b34e1de077ad9c1c : SUCCESS in 39m 23s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/e1aab31ec1834d5fb1aad72a6f6b801a : SUCCESS in 39m 43s","accounts_in_message":[],"_revision_number":2},{"id":"efe2d3f51c11e0b87b5541661982a92ee1e293f3","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-02-11 05:26:53.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"abf530bf3b4a4e7e834a63372aff71aaf30ebbb8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-11 05:27:28.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"242d6705abe4f98a9ef06bae8d7c5889f2726f01","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-11 06:41:10.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d75ebb14b96f42fca73a79e363cb5d25\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e19905f9e4764f9eaf83756604c67de6 : SUCCESS in 4m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8c3e07704b73484386a0cf5ce58b1a29 : SUCCESS in 3m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/20e63d72f87c48328e95d1a91cf72221 : SUCCESS in 4m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9746f439e7914f75a508f134d2498dea : SUCCESS in 4m 00s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/84ab7934839c4d4a8370021fb4cd529a : SUCCESS in 23m 45s\n- aodh-tempest-plugin-mysql-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c2da9148925541bb94318f7427fff0b4 : SUCCESS in 18m 56s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/f4eef8d6ecb240a689c73c6831dcd38f : SUCCESS in 38m 26s (non-voting)\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/b1132c9a849f4cad86761ea9d1936f02 : SUCCESS in 54m 28s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/d627454df6de4961b3f3fbd6c8878941 : SUCCESS in 55m 54s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/33bdf561cbf64071a2f4062c58c1f84d : SUCCESS in 1h 03m 33s","accounts_in_message":[],"_revision_number":2},{"id":"d7a9ab8c464d76528eb8cb654902df574417b0f8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-11 06:41:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"69444d5ce30f1b5fac5d3a762373910582afb331","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-11 06:42:12.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/10ca69f5e2f844139a380078c9d127df\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d440fed7738247249ff2c19c1f768094 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"04f19518e695e76d4e92310c47ea199971e518e4","revisions":{"709c670608d3177120a7efab8d37709df7db3e04":{"kind":"REWORK","_number":1,"created":"2024-11-23 09:00:02.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/78/936078/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/78/936078/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/78/936078/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/78/936078/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/78/936078/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/78/936078/1"}}},"commit":{"parents":[{"commit":"10cde330b5ab0340dbef17bba07df6c4c429c1a8","subject":"reno: Update master for unmaintained/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/10cde330b5ab0340dbef17bba07df6c4c429c1a8"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-23 08:59:25.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-23 08:59:25.000000000","tz":540},"subject":"Drop remaining reference to \u0027test\u0027 extras","message":"Drop remaining reference to \u0027test\u0027 extras\n\nThis was removed when test-requirements was added.\n\nAlso use the common helper to install aodh itself.\n\nChange-Id: I56de124a3a6d4729cfd12d79deab59e96fed9fdd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/709c670608d3177120a7efab8d37709df7db3e04"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/709c670608d3177120a7efab8d37709df7db3e04"}]},"branch":"refs/heads/master"},"04f19518e695e76d4e92310c47ea199971e518e4":{"kind":"REWORK","_number":2,"created":"2024-11-23 09:35:10.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/78/936078/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/78/936078/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/78/936078/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/78/936078/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/78/936078/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/78/936078/2"}}},"commit":{"parents":[{"commit":"10cde330b5ab0340dbef17bba07df6c4c429c1a8","subject":"reno: Update master for unmaintained/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/10cde330b5ab0340dbef17bba07df6c4c429c1a8"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-23 08:59:25.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-23 09:35:02.000000000","tz":540},"subject":"Drop remaining reference to \u0027test\u0027 extras","message":"Drop remaining reference to \u0027test\u0027 extras\n\nThis was removed when test-requirements was added.\n\nAlso use the common helper to install aodh itself.\n\nChange-Id: I56de124a3a6d4729cfd12d79deab59e96fed9fdd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/04f19518e695e76d4e92310c47ea199971e518e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/04f19518e695e76d4e92310c47ea199971e518e4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
