)]}'
{"id":"openstack%2Faodh~936099","triplet_id":"openstack%2Faodh~master~I0a6aff5410c0c49923a83a2ccd850b4919cb62ac","project":"openstack/aodh","branch":"master","topic":"normalize-wsgi","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-03-28 09:42:47.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0a6aff5410c0c49923a83a2ccd850b4919cb62ac","subject":"devstack: Remove mod_wsgi support","status":"MERGED","created":"2024-11-24 13:56:44.000000000","updated":"2025-03-28 09:43:54.000000000","submitted":"2025-03-28 09:42:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"936099-normalize-wsgi","meta_rev_id":"70482b44a6b18b45e111bbe4b6b88e3b44b39e88","_number":936099,"virtual_id_number":936099,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-03-28 09:42:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"all":[{"value":2,"date":"2025-03-24 09:39:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":2,"date":"2025-03-28 08:43:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},{"value":1,"date":"2025-03-25 15:11:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"all":[{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":1,"date":"2025-03-28 08:43:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-24 14:14:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-03-24 09:39:52.000000000","updated_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"reviewer":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"state":"REVIEWER"},{"updated":"2025-03-25 15:11:05.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2025-03-28 08:43:29.000000000","updated_by":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"reviewer":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"state":"REVIEWER"}],"messages":[{"id":"449040799db3b846e3b0a30656d2c946f29e6f37","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-24 13:56:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8cbf0ad7204ce8f84a16bdc1e83fcf4cd1d305ce","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-24 13:56:59.000000000","message":"Topic set to normalize-wsgi","accounts_in_message":[],"_revision_number":1},{"id":"ba2afa22f74f796859c83643911b5334fbb38fb4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-24 13:57:32.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5e739b6b0174c9598fd4b450b462cbc4d59aabea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-24 14:14:43.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e0ef4be35e964f01a0fcb60fe2d4c40a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8858cccdde894f2a9356c4a0a0086b6e : SUCCESS in 3m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b2ed8acfcc844f4e907fbf7dd8d5472a : SUCCESS in 6m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/40ea93f544a746dba64586f24ffdba67 : SUCCESS in 5m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7953fd435b54d9aa62b2b0f28c0faa2 : SUCCESS in 4m 16s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/7689b9f7fbc34b21a08fcfdc612af1d1 : FAILURE in 16m 31s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/1a6880ab40ea4c28bc30a359a70df5d1 : FAILURE in 6m 40s (non-voting)\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/24e27657495f4f408ac972c6ae319bf3 : FAILURE in 9m 14s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/2e92f299fe1c48ac90e3171d8a20a458 : FAILURE in 8m 38s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/c80c9170c79c4344afefa2291ccf453b : FAILURE in 15m 29s","accounts_in_message":[],"_revision_number":2},{"id":"0cf986080d3739d2af9e161004626eb04a4387b5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-24 15:25:45.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"278e35d5b08eee0cfe63cb9abc0117feb8828a59","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-24 16:02:33.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3a9fe4b20a354d638f7229cc7dd54a7d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/72232cc18d334bcc98dcee666a98514d : SUCCESS in 2m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a51b2dc92f4b45d695adf353434de925 : SUCCESS in 3m 07s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9edf35138c404fc48bd489dc9a632ad1 : SUCCESS in 4m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2cf3d95fd03a42a49c1cdd8c0a7b7f78 : SUCCESS in 5m 41s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/8ad6b48b1d4b47f2be2ab6302091ec70 : SUCCESS in 21m 56s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/0012d89d27884a3a9d1737243ab39196 : SUCCESS in 21m 29s (non-voting)\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/63cbeaa35eb844a180473fbb3c7cb2ec : SUCCESS in 26m 12s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/ba1e675e13a245cb9d32786ad8f8b984 : SUCCESS in 31m 38s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/132614adc24448fe8d1563964e1c45d8 : SUCCESS in 35m 46s","accounts_in_message":[],"_revision_number":3},{"id":"6a75386a4e351000f9e73cb4c9b5d570d67d61a6","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2025-03-24 09:39:52.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"cd0866ed8adbaff2f01a58b7f3e7568b1990252d","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-03-25 15:11:05.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"9dff019c89d36220148f7acabded8057ba5de137","author":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"date":"2025-03-28 08:43:29.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"8c66163eda1e4bbae2ebdcbb8d2046d6c9ff133a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-28 08:44:15.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"46fba0134b8531182282be3c110babc56760155e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-28 09:42:47.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/998da7e35af847498371ae87ba8c9e1b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1ca67e9863f4677a9e6fba2eaaf0f69 : SUCCESS in 5m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3e0ee7c34c094fdc9d0d8d2df1c312a1 : SUCCESS in 5m 23s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c2b551d33cab4f85872546c0d3d69685 : SUCCESS in 5m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dfc05f09b4514ccc9327abf70cbf0cf7 : SUCCESS in 8m 44s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/9908f9d6dbb94dafb664b522a214afb6 : SUCCESS in 33m 56s\n- aodh-tempest-plugin-mysql-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7c3f66e048e64aa2addb78c8abf95dbd : SUCCESS in 25m 11s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/3f63e24536434cc8b0a00c304c8cd312 : SUCCESS in 33m 14s (non-voting)\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/a1bb8b92dc76478ca329289909baddaf : SUCCESS in 38m 22s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/6afb1ebc51b149539cf61722e21c1aec : SUCCESS in 47m 08s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/807203f37eba4eb9b7a814d21d78f78b : SUCCESS in 53m 31s","accounts_in_message":[],"_revision_number":3},{"id":"5b373ff6f5894ab90bc65fcf5c0d3acf2db803ea","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-28 09:42:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"70482b44a6b18b45e111bbe4b6b88e3b44b39e88","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-28 09:43:54.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/226c966efdfa4f9b829b64edc86ee167\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/903cecf5771d4d5aa8699833633ba907 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"19369bda050b6af0873610ba6b938754db416f35","revisions":{"802f78d87fc36c5464687728c185f6dfbe71d4cb":{"kind":"REWORK","_number":1,"created":"2024-11-24 13:56:44.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/99/936099/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/99/936099/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/99/936099/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/99/936099/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/99/936099/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/99/936099/1"}}},"commit":{"parents":[{"commit":"5b4f3c8bc8d65830064a402f189c192190480bea","subject":"devstack: Remove support for \"simple\" api service","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/5b4f3c8bc8d65830064a402f189c192190480bea"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-24 13:56:12.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-24 13:56:12.000000000","tz":540},"subject":"devstack: Remove mod_wsgi support","message":"devstack: Remove mod_wsgi support\n\nThe same is being removed from devstack (for example:\nhttps://review.opendev.org/c/openstack/devstack/+/932195 )\n\nChange-Id: I0a6aff5410c0c49923a83a2ccd850b4919cb62ac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/802f78d87fc36c5464687728c185f6dfbe71d4cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/802f78d87fc36c5464687728c185f6dfbe71d4cb"}]},"branch":"refs/heads/master"},"93a77ebc000d0a61e032bcf19f5ccb9b0b0ddf02":{"kind":"REWORK","_number":2,"created":"2024-11-24 13:57:32.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/99/936099/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/99/936099/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/99/936099/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/99/936099/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/99/936099/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/99/936099/2"}}},"commit":{"parents":[{"commit":"5b4f3c8bc8d65830064a402f189c192190480bea","subject":"devstack: Remove support for \"simple\" api service","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/5b4f3c8bc8d65830064a402f189c192190480bea"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-24 13:56:12.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-24 13:57:23.000000000","tz":540},"subject":"devstack: Remove mod_wsgi support","message":"devstack: Remove mod_wsgi support\n\nThe same is being removed from devstack (for example:\nhttps://review.opendev.org/c/openstack/devstack/+/932195 )\n\nChange-Id: I0a6aff5410c0c49923a83a2ccd850b4919cb62ac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/93a77ebc000d0a61e032bcf19f5ccb9b0b0ddf02"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/93a77ebc000d0a61e032bcf19f5ccb9b0b0ddf02"}]},"branch":"refs/heads/master"},"19369bda050b6af0873610ba6b938754db416f35":{"kind":"REWORK","_number":3,"created":"2024-11-24 15:25:45.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/99/936099/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/99/936099/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/99/936099/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/99/936099/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/99/936099/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/99/936099/3"}}},"commit":{"parents":[{"commit":"5b4f3c8bc8d65830064a402f189c192190480bea","subject":"devstack: Remove support for \"simple\" api service","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/5b4f3c8bc8d65830064a402f189c192190480bea"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-24 13:56:12.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-24 15:25:14.000000000","tz":540},"subject":"devstack: Remove mod_wsgi support","message":"devstack: Remove mod_wsgi support\n\nThe same is being removed from devstack (for example:\nhttps://review.opendev.org/c/openstack/devstack/+/932195 )\n\nChange-Id: I0a6aff5410c0c49923a83a2ccd850b4919cb62ac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/19369bda050b6af0873610ba6b938754db416f35"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/19369bda050b6af0873610ba6b938754db416f35"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
