)]}'
{"id":"openstack%2Faodh~936101","triplet_id":"openstack%2Faodh~master~Id7c67ab692105383ad46a5663aff05710485192d","project":"openstack/aodh","branch":"master","topic":"normalize-wsgi","attention_set":{},"removed_from_attention_set":{"5202":{"account":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"last_update":"2025-03-28 09:42:49.000000000","reason":"Change was submitted"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-03-28 09:42:49.000000000","reason":"Change was submitted"},"4264":{"account":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"last_update":"2025-03-24 09:36:42.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"Id7c67ab692105383ad46a5663aff05710485192d","subject":"Use common helper to generate uwsgi setting","status":"MERGED","created":"2024-11-24 14:03:29.000000000","updated":"2025-03-28 09:44:35.000000000","submitted":"2025-03-28 09:42:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"936101-normalize-wsgi","meta_rev_id":"09643b4030783516b009dd8f3248e092650912ca","_number":936101,"virtual_id_number":936101,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-03-28 09:42:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"disliked":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"all":[{"value":-1,"date":"2024-12-03 15:18:09.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":2,"date":"2025-03-24 08:32:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},{"value":2,"date":"2025-03-24 09:36:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":1,"date":"2025-03-25 15:12:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"all":[{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},{"value":1,"date":"2025-03-24 09:36:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-24 14:31:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-12-03 15:18:09.000000000","updated_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"},{"updated":"2025-03-24 08:32:57.000000000","updated_by":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"reviewer":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"state":"REVIEWER"},{"updated":"2025-03-24 08:32:57.000000000","updated_by":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"reviewer":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"state":"REVIEWER"},{"updated":"2025-03-25 15:12:24.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"9830bcb42880e3374b0ceb8cab2576d0ee908c52","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-24 14:03:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"45e0d968123c94577aa4f97c5cfd99ba58a5bc0b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-24 14:31:29.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/82a050ff96714b6abf6c05779fc67c75\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa296affa0464f84a1f11d85b10719b0 : SUCCESS in 4m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5994b9a082ae492b90d5e4fd449bed26 : SUCCESS in 5m 31s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5deb1b226bc54b7195b19a92944d9f0c : SUCCESS in 6m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14f3f2b96b8f4e4da90fa21b27b1bc32 : SUCCESS in 6m 13s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/0d48f07de1b348b5aa624d1b3e8433b6 : FAILURE in 25m 27s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/08aa5ba839204790a4ae82c77996b4ba : FAILURE in 16m 21s (non-voting)\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/90aa231670224f9c958b1ed902d3ec38 : FAILURE in 24m 56s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/fbcd5f267fe8407b8739f8ee7bb0d918 : FAILURE in 21m 27s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/74ce3e1d9d2b4273b40ff147b17e55c3 : FAILURE in 23m 05s","accounts_in_message":[],"_revision_number":1},{"id":"567c46157e346509c346f8e2164945235e0a6252","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-24 15:25:45.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"a31cf4a80e046889777a25152eca2cabb5dc9bc3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-24 16:06:51.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c35348f1ce4c48c093e3aa7545905b01\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af94605e86fa41cca533d415a8bdb234 : SUCCESS in 3m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/530c81d5758b44a3b093e07c98eb9a9a : SUCCESS in 3m 13s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b616a88efd784a5e8b9a4f3485c428b9 : SUCCESS in 3m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a64c795a6fe4370b0f8fa90a3d03950 : SUCCESS in 3m 58s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/6f8e497938024c89b23ec82ebc758e99 : FAILURE in 17m 18s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/78b55cf232fd48cb90b50c88a259b09d : FAILURE in 15m 40s (non-voting)\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/49199d6eea1d43579c7fec30a0766e4e : FAILURE in 17m 32s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/feaafa867e684dca9343c0b035427955 : FAILURE in 32m 25s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/f20a7c4282b64b9396d68b9662b20a37 : FAILURE in 25m 51s","accounts_in_message":[],"_revision_number":2},{"id":"95015d167d920290bd8ae0e80c03a76a7f944a9e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-24 16:48:53.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"1071c5ac73d776804a4db1835025b0590c21dee0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-24 16:53:50.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"29e13b48576708e1422cece3b46c46eb905eb70d","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-24 16:57:46.000000000","message":"Topic set to normalize-wsgi","accounts_in_message":[],"_revision_number":4},{"id":"ec6af30ece964012c50f041798164f7e2710d1b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-24 17:49:52.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a0eddd2e3090479e89acd8a709b9e3e5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/754288433e014a1e806d8f79ce70b5e3 : SUCCESS in 3m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7d2b769ad97a4821b09827d830ad980c : SUCCESS in 4m 11s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6075dd969361495092904e7fab64970b : SUCCESS in 4m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2e35fd22cf549d0b9ee9cf44fe48df3 : SUCCESS in 5m 04s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/ddded3411e2f4c1c9cc6937d6461b522 : SUCCESS in 39m 58s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/91f5f75c1ae9445e9f055885a28a507f : SUCCESS in 37m 06s (non-voting)\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/7c077e957105490c8a61564329818885 : SUCCESS in 26m 00s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/03847a035c304c11a1d996ba7cc370f6 : SUCCESS in 50m 51s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/08d15b8d27c545a88957b62e612c710f : SUCCESS in 46m 32s","accounts_in_message":[],"_revision_number":4},{"id":"4c082aabb29ff38aaa78a06d9c839d02a2f4602e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-24 18:27:36.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"bd9a3e853f2e39c8f848fddb1a34ba12a1ef199f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-24 19:07:23.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3f68f3a16a044c3498fa4c1bb2c8135c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c90f07a7aadf48278a316b439d0ba069 : SUCCESS in 2m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/924550b6eb68459cbe574a31a50e78d2 : SUCCESS in 3m 38s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5c9a17401c0442299cb028429165fc78 : SUCCESS in 3m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73fd9827840b494cbe8545b60fad5f36 : SUCCESS in 5m 34s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/80cf806b73844b9fbe8730b3993e90e9 : SUCCESS in 34m 14s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/17f27371e0a646cd83af738d98a647b2 : SUCCESS in 19m 28s (non-voting)\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/46485449b16d4720b98cb46b88b3f730 : SUCCESS in 25m 37s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/e6351108e80c44059a7ddc620e9cb613 : SUCCESS in 31m 47s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/c0f794c7e2da4ff3bd7708dd3b2a614b : SUCCESS in 34m 35s","accounts_in_message":[],"_revision_number":5},{"id":"3758c6ded4e9de35bc24519ff47d771c247ac152","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2024-12-03 15:18:09.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"8ebc4dd59ef8fa2a33eb186afb02f43a9d6063ee","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-12-04 06:54:30.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"ee47b8cd99d6fe3428ba3c62eabe74c484f502c0","author":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"date":"2025-03-24 08:32:57.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"a63ee4250db0f54850df2acadb31b556e84b62b7","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2025-03-24 09:36:42.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"ad282d7e26d0fad7b7aa7cb25ac143667734d5ee","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2025-03-24 09:36:46.000000000","message":"Patch Set 5: Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"98971ccc411519b6e74ece43c2aec97c731dcf38","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-03-25 15:12:24.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"217d65f40ca2d0beefb5cc0ee4d41a93a3ee57bb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-28 08:44:18.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"0f8667daa0bf8cd858078f9c9b0972d93b7d8f75","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-28 09:42:48.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9eb34c09f8b74096919500a6ba494338\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/66477e26480d4556b24beb91640c576a : SUCCESS in 4m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d01674337e4b46e3b02b40fc353feea8 : SUCCESS in 4m 26s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ab40897083154baab089f8cd7e70638d : SUCCESS in 5m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ecbd2f4b1ee4c32ad075ab57142c18c : SUCCESS in 8m 45s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/70e32c911aa54ef38256212e5978f88c : SUCCESS in 33m 16s\n- aodh-tempest-plugin-mysql-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ac2cdfcc69db4805bfd478ef87b6d2bc : SUCCESS in 30m 01s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/542deb7ba5f64182aab556e2693627ad : SUCCESS in 36m 19s (non-voting)\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/0d7759f38ddc4e18bd3ebd79827fa3bc : SUCCESS in 43m 30s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/7d65025a633145a99b0b4bfb19e95360 : SUCCESS in 43m 50s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/1f86279003b54cbd954febcf3cd3e9c0 : SUCCESS in 45m 28s","accounts_in_message":[],"_revision_number":5},{"id":"9a543836680b461178acda38e166b08df3191789","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-28 09:42:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"09643b4030783516b009dd8f3248e092650912ca","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-28 09:44:35.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ed061c18c9b14efca3446d30e1d1b721\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9db24e5f981f44fda93ce1b349463caa : SUCCESS in 34s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"97e7362624191adeb0407d8dbf1e1c7ba8d6a3cd","revisions":{"134c4dce882cd9f3672ec95ea32dbf912fa755aa":{"kind":"REWORK","_number":1,"created":"2024-11-24 14:03:29.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/01/936101/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/01/936101/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/01/936101/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/01/936101/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/01/936101/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/01/936101/1"}}},"commit":{"parents":[{"commit":"93a77ebc000d0a61e032bcf19f5ccb9b0b0ddf02","subject":"devstack: Remove mod_wsgi support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/93a77ebc000d0a61e032bcf19f5ccb9b0b0ddf02"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-24 14:02:52.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-24 14:02:52.000000000","tz":540},"subject":"Use common helper to generate uwsgi setting","message":"Use common helper to generate uwsgi setting\n\nChange-Id: Id7c67ab692105383ad46a5663aff05710485192d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/134c4dce882cd9f3672ec95ea32dbf912fa755aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/134c4dce882cd9f3672ec95ea32dbf912fa755aa"}]},"branch":"refs/heads/master"},"c1217af9aebb33ffde7c677c756ce906843899c9":{"kind":"REWORK","_number":2,"created":"2024-11-24 15:25:45.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/01/936101/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/01/936101/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/01/936101/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/01/936101/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/01/936101/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/01/936101/2"}}},"commit":{"parents":[{"commit":"19369bda050b6af0873610ba6b938754db416f35","subject":"devstack: Remove mod_wsgi support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/19369bda050b6af0873610ba6b938754db416f35"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-24 14:02:52.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-24 15:25:30.000000000","tz":540},"subject":"Use common helper to generate uwsgi setting","message":"Use common helper to generate uwsgi setting\n\nChange-Id: Id7c67ab692105383ad46a5663aff05710485192d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/c1217af9aebb33ffde7c677c756ce906843899c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/c1217af9aebb33ffde7c677c756ce906843899c9"}]},"branch":"refs/heads/master"},"8c244e4dc586ea39871cb083283160107bf4b5d7":{"kind":"REWORK","_number":3,"created":"2024-11-24 16:48:53.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/01/936101/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/01/936101/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/01/936101/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/01/936101/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/01/936101/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/01/936101/3"}}},"commit":{"parents":[{"commit":"19369bda050b6af0873610ba6b938754db416f35","subject":"devstack: Remove mod_wsgi support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/19369bda050b6af0873610ba6b938754db416f35"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-24 14:02:52.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-24 16:48:43.000000000","tz":540},"subject":"Use common helper to generate uwsgi setting","message":"Use common helper to generate uwsgi setting\n\nChange-Id: Id7c67ab692105383ad46a5663aff05710485192d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/8c244e4dc586ea39871cb083283160107bf4b5d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/8c244e4dc586ea39871cb083283160107bf4b5d7"}]},"branch":"refs/heads/master"},"6f4f669723122a1f845f2c29d0e76b2bc8381385":{"kind":"REWORK","_number":4,"created":"2024-11-24 16:53:50.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/01/936101/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/01/936101/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/01/936101/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/01/936101/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/01/936101/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/01/936101/4"}}},"commit":{"parents":[{"commit":"19369bda050b6af0873610ba6b938754db416f35","subject":"devstack: Remove mod_wsgi support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/19369bda050b6af0873610ba6b938754db416f35"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-24 14:02:52.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-24 16:52:48.000000000","tz":540},"subject":"Use common helper to generate uwsgi setting","message":"Use common helper to generate uwsgi setting\n\n... to reduce own logic in our devstack plugin. Note that this adds\napache web server in front of aodh-api run by uwsgi, following\nthe standard architecture in devstack.\n\nChange-Id: Id7c67ab692105383ad46a5663aff05710485192d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/6f4f669723122a1f845f2c29d0e76b2bc8381385"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/6f4f669723122a1f845f2c29d0e76b2bc8381385"}]},"branch":"refs/heads/master"},"97e7362624191adeb0407d8dbf1e1c7ba8d6a3cd":{"kind":"REWORK","_number":5,"created":"2024-11-24 18:27:36.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/01/936101/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/01/936101/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/01/936101/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/01/936101/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/01/936101/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/01/936101/5"}}},"commit":{"parents":[{"commit":"19369bda050b6af0873610ba6b938754db416f35","subject":"devstack: Remove mod_wsgi support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/19369bda050b6af0873610ba6b938754db416f35"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-24 14:02:52.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-24 18:27:25.000000000","tz":540},"subject":"Use common helper to generate uwsgi setting","message":"Use common helper to generate uwsgi setting\n\n... to reduce own logic in our devstack plugin. Note that this adds\napache web server in front of aodh-api run by uwsgi, following\nthe standard architecture in devstack.\n\nChange-Id: Id7c67ab692105383ad46a5663aff05710485192d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/97e7362624191adeb0407d8dbf1e1c7ba8d6a3cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/97e7362624191adeb0407d8dbf1e1c7ba8d6a3cd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
