)]}'
{"id":"openstack%2Faodh~936211","triplet_id":"openstack%2Faodh~master~I7ce471c7fbce960ba92bfd18203a73454088314e","project":"openstack/aodh","branch":"master","topic":"normalize-wsgi","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-03-28 09:45:06.000000000","reason":"Change was submitted"},"4264":{"account":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"last_update":"2025-03-25 16:30:09.000000000","reason":"\u003cGERRIT_ACCOUNT_4264\u003e replied on the change","reason_account":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"}},"32968":{"account":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"last_update":"2025-03-25 16:09:56.000000000","reason":"removed on reply"},"34975":{"account":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"last_update":"2025-03-28 09:45:06.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I7ce471c7fbce960ba92bfd18203a73454088314e","subject":"Add wsgi module","status":"MERGED","created":"2024-11-26 01:51:26.000000000","updated":"2025-03-28 09:45:57.000000000","submitted":"2025-03-28 09:45:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"936211-normalize-wsgi","meta_rev_id":"a6c7812a2873737890da10cac0135f573779cf3f","_number":936211,"virtual_id_number":936211,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-03-28 09:45:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"all":[{"value":2,"date":"2025-03-25 16:09:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},{"value":2,"date":"2025-03-25 16:30:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":1,"date":"2025-03-25 15:13:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"all":[{"value":1,"date":"2025-03-28 08:32:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},{"value":1,"date":"2025-03-25 16:30:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-26 03:45:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-03-24 08:30:49.000000000","updated_by":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"reviewer":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"state":"REVIEWER"},{"updated":"2025-03-24 08:31:09.000000000","updated_by":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"reviewer":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"state":"REVIEWER"},{"updated":"2025-03-25 10:17:01.000000000","updated_by":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"reviewer":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"state":"REVIEWER"},{"updated":"2025-03-25 15:13:13.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"b59bf0cf25d85d49965c46470b27b633ecf19b61","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-26 01:51:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e4b9ee9e2bc774147f9d88c730527a6dd753e3b8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-26 02:35:46.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"afc83d0adb24532bfe91058f2bbea4c6eb69b595","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-26 03:45:59.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b093a28d71444031b5cfe2ed13f9bdd0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56f4fbf246d6460c8dcca6cc495d7e9f : SUCCESS in 3m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/36c46a7c13da45748874a7aca5a87dd3 : SUCCESS in 5m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9ca90505ef7848c3882b7504369eb3d0 : SUCCESS in 5m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc8b4e48c78041f3a0211013807657a8 : SUCCESS in 8m 47s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/70986806656c456387074672270b4a90 : SUCCESS in 36m 50s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/21534746a3b94c36be589db767c283de : SUCCESS in 36m 39s (non-voting)\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/867ff54367314b568a2b09e97ec3418f : SUCCESS in 43m 25s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/b31f617181de450b888bf36b61035780 : SUCCESS in 56m 50s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/edfc9a311e60433da0b9476d871d1d23 : SUCCESS in 57m 15s","accounts_in_message":[],"_revision_number":2},{"id":"0b811598617aa9a1a69a4193a704004708e7a614","author":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"date":"2025-03-24 08:30:49.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"9ad2596dc5b6061f9a96017a309eb8e02809794c","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2025-03-24 09:37:54.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"b789c062c2220aa72fbaeff7251b26ad32bc33a9","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2025-03-24 09:37:58.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"e7dd837b2aefe2289d6fa48cefba9790d20522a9","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2025-03-25 10:17:01.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d7d78ee05973d9676919d65aa176dece6b29d8c0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-03-25 10:43:41.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n* Workflow+1\n","accounts_in_message":[],"_revision_number":3},{"id":"7277e3458818130da36660c18f0b0ada89aa3282","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-03-25 10:43:48.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"64895b6e3720278cd9104af360885f9289f275b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-25 11:35:33.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8b04029fd23c43feb1dc4104506fa8af\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6598743582fd49dc8f8eab7cd4b65c44 : SUCCESS in 5m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/13942c9962d34a53b821c736c2381f84 : SUCCESS in 5m 37s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/bd425188f9ae4a78b0a9d3df7dd96218 : SUCCESS in 5m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/536c5a239397469988b39cd64d9af6e0 : SUCCESS in 8m 06s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/c238dede24804c9793820568be164cbd : SUCCESS in 30m 23s\n- aodh-tempest-plugin-mysql-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d9cb8177bdc4460b868c4f48c9253971 : SUCCESS in 30m 05s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/8b2f1abd2e9a4968b24f6ea8ad1253fb : SUCCESS in 29m 59s (non-voting)\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/37e4ce5aafb141229624b858a724bcbc : SUCCESS in 45m 18s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/0241b05eeb4f4a81843106147fc91613 : SUCCESS in 27m 20s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/6245ac3e2aee415087f68e5094d3b202 : FAILURE in 27m 51s","accounts_in_message":[],"_revision_number":3},{"id":"7b02e25b9d7549b6c82da4fa8dd1bab1ee3c7514","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-03-25 12:05:20.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"5ce8c687ca7d47bd9493d0e4efe83e6a6441d648","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-25 12:58:19.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/45249394ca5549c8b80fbeb2b030ba7e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d64f2cc9fffe47559c480a8612165486 : SUCCESS in 2m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bd7da8368579477b97835824eb075e63 : SUCCESS in 5m 30s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fda9a49692684334b367623a465f4e6a : SUCCESS in 6m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66c54805224c417ea46486d60e51ac50 : SUCCESS in 9m 49s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/57afa39344934b5d9acff6829617e63d : SUCCESS in 33m 22s\n- aodh-tempest-plugin-mysql-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4027bbcb6f264d72b3add91203f38be5 : SUCCESS in 30m 04s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/4066a528e5a64d67a616a801a40592c3 : SUCCESS in 19m 14s (non-voting)\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/8368fd84ad2c4914bcead89ab9764921 : SUCCESS in 48m 40s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/d13d8ffd31004f4e8eb48d765496821e : SUCCESS in 21m 43s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/20d160c6d0544f42a18e14d3cfb2ba1e : SUCCESS in 48m 26s","accounts_in_message":[],"_revision_number":3},{"id":"6ef570577447b85e07400eaa5bf2e9eb7cfeab8b","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-03-25 15:13:13.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"cf90ce3e5d7ea631ba8bdd1ee49ad51cfefa6cfb","author":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"date":"2025-03-25 16:09:56.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"73dd661e29c816ac9c1c44ad4f28cef9b1904b2d","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2025-03-25 16:30:09.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"20413989d3ace07211ad64ccd2e296627fc4e153","author":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"date":"2025-03-28 08:32:14.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"f65fc724f085f8ce9f23f741dcaa206e0221678a","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2025-03-28 08:35:38.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"4a0f6b4cedbed5d3687dfa0ed6353d4765727ff9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-28 08:44:20.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"2110af37cb0dcf2259b298eddeae7979cbafb589","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-28 09:45:05.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f0d00244953942faaebbfefba77a79c5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/96146970eb234e86be4c3e0d0b482db5 : SUCCESS in 5m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/411ee3fffcda40f59a560cdb5318d395 : SUCCESS in 2m 35s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/99eb61bf7bc7466687dc904c1b854dd6 : SUCCESS in 4m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fbd554228e0347d7b99cf674d8e88752 : SUCCESS in 9m 16s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/0bfec7370c0846d8be14499895a27543 : SUCCESS in 32m 30s\n- aodh-tempest-plugin-mysql-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/83f26f69c37948819dc2399b324b0220 : SUCCESS in 35m 46s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/3d20490af1bf4b7391cfc08f4022fc8d : SUCCESS in 14m 35s (non-voting)\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/2b2b191c222b45f5aac90560b54127b2 : SUCCESS in 40m 48s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/6a017707ff174b9e96dc5c61ddc8151a : SUCCESS in 43m 21s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/3adb8f164a82479992955b04843e4846 : SUCCESS in 49m 38s","accounts_in_message":[],"_revision_number":3},{"id":"332f174bc3603f580e0dfd31abc1cca808980293","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-28 09:45:06.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"a6c7812a2873737890da10cac0135f573779cf3f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-28 09:45:57.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/573a259ed1654e9aaba55e4adeb6e917\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/462cf54375c949f1b256ae2c3a8c1ac1 : SUCCESS in 32s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"f7aa502e3c62e8fd599e06bdd870a20e8437c90b","revisions":{"8ae988ea4acb76c3028ecc69acde2b8ed0ffe2fe":{"kind":"REWORK","_number":1,"created":"2024-11-26 01:51:26.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/11/936211/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/11/936211/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/11/936211/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/11/936211/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/11/936211/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/11/936211/1"}}},"commit":{"parents":[{"commit":"97e7362624191adeb0407d8dbf1e1c7ba8d6a3cd","subject":"Use common helper to generate uwsgi setting","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/97e7362624191adeb0407d8dbf1e1c7ba8d6a3cd"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-26 01:50:32.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-26 01:50:32.000000000","tz":540},"subject":"Add wsgi module","message":"Add wsgi module\n\nThe wsgi_script feature is being removed because of some changes in\nunderlying python packaging tooling. This makes aodh to vendor the wsgi\nmodule which can be used instead of the wsgi script, according to\nthe proposed community goal[1]. The existing wsgi scripts are kept now\nfor smooth transition.\n\nThe exsiting app.wsgi file is deprecated in favor of the new wsgi\nmodule to follow the standard pattern to vendor the wsgi module.\n\n[1] https://review.opendev.org/c/openstack/governance/+/902807\n\nChange-Id: I7ce471c7fbce960ba92bfd18203a73454088314e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/8ae988ea4acb76c3028ecc69acde2b8ed0ffe2fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/8ae988ea4acb76c3028ecc69acde2b8ed0ffe2fe"}]},"branch":"refs/heads/master"},"9985590e982a6397b4b7a0317746008506fd5e14":{"kind":"REWORK","_number":2,"created":"2024-11-26 02:35:46.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/11/936211/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/11/936211/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/11/936211/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/11/936211/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/11/936211/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/11/936211/2"}}},"commit":{"parents":[{"commit":"97e7362624191adeb0407d8dbf1e1c7ba8d6a3cd","subject":"Use common helper to generate uwsgi setting","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/97e7362624191adeb0407d8dbf1e1c7ba8d6a3cd"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-26 01:50:32.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-26 02:35:34.000000000","tz":540},"subject":"Add wsgi module","message":"Add wsgi module\n\nThe wsgi_script feature is being removed because of some changes in\nunderlying python packaging tooling. This makes aodh to vendor the wsgi\nmodule which can be used instead of the wsgi script, according to\nthe proposed community goal[1]. The existing wsgi scripts are kept now\nfor smooth transition.\n\nThe exsiting app.wsgi file is deprecated in favor of the new wsgi\nmodule to follow the standard pattern to vendor the wsgi module.\n\n[1] https://review.opendev.org/c/openstack/governance/+/902807\n\nChange-Id: I7ce471c7fbce960ba92bfd18203a73454088314e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/9985590e982a6397b4b7a0317746008506fd5e14"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/9985590e982a6397b4b7a0317746008506fd5e14"}]},"branch":"refs/heads/master"},"f7aa502e3c62e8fd599e06bdd870a20e8437c90b":{"kind":"REWORK","_number":3,"created":"2025-03-25 10:43:41.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/11/936211/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/11/936211/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/11/936211/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/11/936211/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/11/936211/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/11/936211/3"}}},"commit":{"parents":[{"commit":"97e7362624191adeb0407d8dbf1e1c7ba8d6a3cd","subject":"Use common helper to generate uwsgi setting","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/97e7362624191adeb0407d8dbf1e1c7ba8d6a3cd"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-26 01:50:32.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-03-25 10:43:41.000000000","tz":0},"subject":"Add wsgi module","message":"Add wsgi module\n\nThe wsgi_script feature is being removed because of some changes in\nunderlying python packaging tooling. This makes aodh to vendor the wsgi\nmodule which can be used instead of the wsgi script, according to\nthe proposed community goal[1]. The existing wsgi scripts are kept now\nfor smooth transition.\n\nThe exsiting app.wsgi file is deprecated in favor of the new wsgi\nmodule to follow the standard pattern to vendor the wsgi module.\n\n[1] https://review.opendev.org/c/openstack/governance/+/902807\n\nChange-Id: I7ce471c7fbce960ba92bfd18203a73454088314e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/f7aa502e3c62e8fd599e06bdd870a20e8437c90b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/f7aa502e3c62e8fd599e06bdd870a20e8437c90b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
