)]}'
{"id":"openstack%2Faodh~936259","triplet_id":"openstack%2Faodh~master~If30adba443ead164af55c947a1822f8690da5e53","project":"openstack/aodh","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-12-10 10:43:40.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If30adba443ead164af55c947a1822f8690da5e53","subject":"Drop logic for absence of segregation rule","status":"MERGED","created":"2024-11-26 13:24:22.000000000","updated":"2024-12-10 10:43:40.000000000","submitted":"2024-12-10 10:43:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"936259","meta_rev_id":"a98ad3cd2fd2f99456f89993cf5d31ae6d27753d","_number":936259,"virtual_id_number":936259,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-12-10 10:43:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"all":[{"value":2,"date":"2024-12-04 13:45:11.000000000","_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":2,"date":"2024-12-05 07:00:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"all":[{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":1,"date":"2024-12-05 07:00:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-26 16:22:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-12-04 13:45:11.000000000","updated_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"},{"updated":"2024-12-05 07:00:39.000000000","updated_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"reviewer":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"state":"REVIEWER"}],"messages":[{"id":"d9fbf43abde72b3ee6c489df734a4465d2d1d063","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-26 13:24:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fff6738e86682161d943433038aa571355a535e0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-26 14:22:35.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"c379ef99a4af80c146eee4ef30eaa09bfea9c26c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-26 14:31:51.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"8c8e07d3c81046db0ed7ca52dc987b1e5e6502ef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-26 14:47:23.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"b303773aff6ada6114229a042d1b0245a44f5510","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-26 15:33:53.000000000","message":"Patch Set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"9e17f23b0d87ed74082ac6d8e0b065f239f986ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-26 16:22:13.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ed9e94290f48459cb1556e3ae407d016\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89b91b6b92e340b680b31707ffe5ef38 : SUCCESS in 5m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/72b940e09c484370a7914a262fe600ee : SUCCESS in 6m 35s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/34cd85881bef44c39014edd3bf5dba1a : SUCCESS in 9m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/138043f9ffaf47f5940dd04666aab37c : SUCCESS in 8m 26s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/3707da471ba646bab60a406286f81852 : SUCCESS in 39m 27s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/08acbe0f72f147f9ac728603d04a17a0 : SUCCESS in 36m 07s (non-voting)\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/ff14028c62e54ee2a34941b7c4e1cd40 : SUCCESS in 43m 24s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/26518491c7b3493ebc9167da63d2c78f : SUCCESS in 39m 21s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/16175e9b4b3d4be1afcc70d0785e72f3 : SUCCESS in 46m 25s","accounts_in_message":[],"_revision_number":5},{"id":"1a4992f65cca84d7189efca08b63841b70e94124","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2024-12-04 13:45:11.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"310c074951590a3cd33e140eabd881858f0f9975","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2024-12-05 07:00:39.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"97ed838df64c121049c738c4d1cc2708617bcb91","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2024-12-05 07:00:44.000000000","message":"Patch Set 5: Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"d47336fe1326f6621fa073d822aed2f083d489dd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-10 09:34:13.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"57b12b9735391896793b00400ff2211b2107d63f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-10 10:43:40.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6141d8f62941418c832e2814be579f97\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e2ec51fb80bd457cbeec8ce1939abfa9 : SUCCESS in 4m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ed539f48a5b6421fab709db441384c50 : SUCCESS in 4m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ee93294a1a3645f99100bdf8ad2fc263 : SUCCESS in 6m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3d45238829e459698ed2cea957cebb1 : SUCCESS in 6m 58s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/8c634bbbbe804f85bab4b1f6cc8c0e44 : SUCCESS in 41m 09s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/47d44d7e40a341c1a1ebd92b34a2cdbc : SUCCESS in 20m 08s (non-voting)\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/45e347d1934f4347a87de109cd3c6749 : SUCCESS in 46m 21s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/62ea0c46c6ac4b4092ccaf182059d5e1 : SUCCESS in 44m 54s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/b51d9c379b5f45d89a766e3a715393a0 : SUCCESS in 35m 20s","accounts_in_message":[],"_revision_number":5},{"id":"a98ad3cd2fd2f99456f89993cf5d31ae6d27753d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-10 10:43:40.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"fac8aa64a0468f0c1bba3ad8e0a45623e7ad48db","revisions":{"756a0ff950a2039911887472c2a6f732c4cfe3dd":{"kind":"REWORK","_number":1,"created":"2024-11-26 13:24:22.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/59/936259/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/59/936259/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/59/936259/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/59/936259/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/59/936259/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/59/936259/1"}}},"commit":{"parents":[{"commit":"10cde330b5ab0340dbef17bba07df6c4c429c1a8","subject":"reno: Update master for unmaintained/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/10cde330b5ab0340dbef17bba07df6c4c429c1a8"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-26 13:23:21.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-26 13:24:10.000000000","tz":540},"subject":"Drop compatibility code with polocy with segregation rule","message":"Drop compatibility code with polocy with segregation rule\n\nSince all policy rules were implemented as code, there is no chance\nthat the seggregation policy rule does not exist.\n\nChange-Id: If30adba443ead164af55c947a1822f8690da5e53\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/756a0ff950a2039911887472c2a6f732c4cfe3dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/756a0ff950a2039911887472c2a6f732c4cfe3dd"}]},"branch":"refs/heads/master"},"f4370198fcca10bd8186c5cb2e5a350c57f5b704":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2024-11-26 14:22:35.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/59/936259/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/59/936259/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/59/936259/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/59/936259/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/59/936259/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/59/936259/2"}}},"commit":{"parents":[{"commit":"10cde330b5ab0340dbef17bba07df6c4c429c1a8","subject":"reno: Update master for unmaintained/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/10cde330b5ab0340dbef17bba07df6c4c429c1a8"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-26 13:23:21.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-26 14:22:35.000000000","tz":0},"subject":"Drop logic for absense of segregation rule","message":"Drop logic for absense of segregation rule\n\nSince all policy rules were implemented as code, there is no chance\nthat the seggregation policy rule does not exist.\n\nChange-Id: If30adba443ead164af55c947a1822f8690da5e53\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/f4370198fcca10bd8186c5cb2e5a350c57f5b704"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/f4370198fcca10bd8186c5cb2e5a350c57f5b704"}]},"branch":"refs/heads/master","description":"Edit commit message"},"f865a96cf7664ffc657a08c0bd56adc5e10a4258":{"kind":"REWORK","_number":3,"created":"2024-11-26 14:31:51.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/59/936259/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/59/936259/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/59/936259/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/59/936259/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/59/936259/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/59/936259/3"}}},"commit":{"parents":[{"commit":"2813c83e5b06a49dc2fbc72917df06bce03680db","subject":"Load full contexts from request headers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/2813c83e5b06a49dc2fbc72917df06bce03680db"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-26 13:23:21.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-26 14:24:43.000000000","tz":540},"subject":"Drop logic for absense of segregation rule","message":"Drop logic for absense of segregation rule\n\nSince all policy rules were implemented as code, there is no chance\nthat the seggregation policy rule does not exist.\n\nChange-Id: If30adba443ead164af55c947a1822f8690da5e53\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/f865a96cf7664ffc657a08c0bd56adc5e10a4258"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/f865a96cf7664ffc657a08c0bd56adc5e10a4258"}]},"branch":"refs/heads/master"},"6a529bbc40e17401f9aaa92dca60532dfede90aa":{"kind":"REWORK","_number":4,"created":"2024-11-26 14:47:23.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/59/936259/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/59/936259/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/59/936259/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/59/936259/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/59/936259/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/59/936259/4"}}},"commit":{"parents":[{"commit":"2813c83e5b06a49dc2fbc72917df06bce03680db","subject":"Load full contexts from request headers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/2813c83e5b06a49dc2fbc72917df06bce03680db"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-26 13:23:21.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-26 14:44:51.000000000","tz":540},"subject":"Drop logic for absense of segregation rule","message":"Drop logic for absense of segregation rule\n\nSince all policy rules were implemented as code, there is no chance\nthat the seggregation policy rule does not exist.\n\nChange-Id: If30adba443ead164af55c947a1822f8690da5e53\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/6a529bbc40e17401f9aaa92dca60532dfede90aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/6a529bbc40e17401f9aaa92dca60532dfede90aa"}]},"branch":"refs/heads/master"},"fac8aa64a0468f0c1bba3ad8e0a45623e7ad48db":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2024-11-26 15:33:53.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/59/936259/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/59/936259/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/59/936259/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/59/936259/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/59/936259/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/59/936259/5"}}},"commit":{"parents":[{"commit":"2813c83e5b06a49dc2fbc72917df06bce03680db","subject":"Load full contexts from request headers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/2813c83e5b06a49dc2fbc72917df06bce03680db"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-26 13:23:21.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-26 15:33:53.000000000","tz":0},"subject":"Drop logic for absence of segregation rule","message":"Drop logic for absence of segregation rule\n\nSince all policy rules were implemented as code, there is no chance\nthat the segregation policy rule does not exist.\n\nChange-Id: If30adba443ead164af55c947a1822f8690da5e53\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/fac8aa64a0468f0c1bba3ad8e0a45623e7ad48db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/fac8aa64a0468f0c1bba3ad8e0a45623e7ad48db"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
