)]}'
{"id":"openstack%2Faodh~936268","triplet_id":"openstack%2Faodh~master~I0e9a12670b8df448bed97448f8de9e3bbf207364","project":"openstack/aodh","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-12-10 10:43:41.000000000","reason":"Change was submitted"},"4264":{"account":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"last_update":"2024-12-06 06:53:44.000000000","reason":"\u003cGERRIT_ACCOUNT_4264\u003e replied on the change","reason_account":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"}}},"hashtags":[],"change_id":"I0e9a12670b8df448bed97448f8de9e3bbf207364","subject":"Add policy rule for get quotas","status":"MERGED","created":"2024-11-26 14:47:23.000000000","updated":"2024-12-10 10:45:34.000000000","submitted":"2024-12-10 10:43:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"936268","meta_rev_id":"bd61e731679d62c30d474e9e40150fe4507495e2","_number":936268,"virtual_id_number":936268,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-12-10 10:43:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"all":[{"value":2,"date":"2024-12-04 13:50:44.000000000","_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":2,"date":"2024-12-06 06:53:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"all":[{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":1,"date":"2024-12-06 06:53:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-26 16:27:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-12-04 13:50:44.000000000","updated_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"},{"updated":"2024-12-05 07:03:58.000000000","updated_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"reviewer":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"state":"CC"},{"updated":"2024-12-06 06:53:44.000000000","updated_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"reviewer":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"state":"REVIEWER"}],"messages":[{"id":"065e4f2e1c5e23353e19bb9dea2ca609aec3ad32","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-26 14:47:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ce0fbd26012258d17a82dc7939c81973d1e9af3d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-26 15:34:00.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"944ce174e87e59804a9f971a71caef13b9aac7ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-26 16:27:19.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1b8d6269938e4fa18639d7804519f654\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3180f5bb61c7460d89d7fd92a2d94b0f : SUCCESS in 4m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/788568e9a02f4163b44fa5e3b4edf9f5 : SUCCESS in 4m 39s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ce0ee62223bb4eeb97838bd7c5c64d7d : SUCCESS in 5m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/359e24b82862417885e3c30f87c59d50 : SUCCESS in 7m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bf26df5f3d794d1da56c9ed548c7b124 : SUCCESS in 7m 11s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/7baaa8939ad1413783f1718e8a281958 : SUCCESS in 25m 54s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/938d639b8026410db377859e5f815f4a : SUCCESS in 19m 04s (non-voting)\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/b65574d6d0e242abb9c1ad3e29ed1cb5 : SUCCESS in 52m 27s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/0f551b090652409ca78662a838b34314 : SUCCESS in 44m 47s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/799906eba2ac42b3bd09c5145afbe07f : SUCCESS in 42m 15s","accounts_in_message":[],"_revision_number":2},{"id":"b39478e43a27430c3e9e8c76750d1760c94b1904","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2024-12-04 13:50:44.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c8009a6c39df2c8548eee1c1b0a367f9bc219aa3","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2024-12-05 07:03:58.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0a14ff7149e5f29d77edcf2854aa09618ed18962","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-12-06 05:08:51.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2eccb50f7e0f9eec04145b5d38a108b6c3c5f460","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2024-12-06 06:53:44.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a9d800de3ecf0c1ae73ec35f7764cc74ebc102dd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-10 09:34:16.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9eb617ba0050bc2e39932b12f7e566c372c03725","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-10 10:43:41.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1200ab241c0d4a0a80a5a4689ea765be\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/754dd5aa7db84b99b46f57be583d96e2 : SUCCESS in 6m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c5c7b5c1f687411186c5f3533cb65deb : SUCCESS in 6m 40s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f79ce3987711413493bd048ff352eccf : SUCCESS in 7m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c31fa9c464f34ab28421f7ec486d812e : SUCCESS in 7m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/69f89c9b6acd4cb2b672dc602c9ecee1 : SUCCESS in 6m 29s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/d44c1a885b4a40818106ef66966499d8 : SUCCESS in 42m 16s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/2be3156eb9de4b55a88f89b2d3d5bd0c : SUCCESS in 30m 49s (non-voting)\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/1cf4cc5270b44ef5a65012a9e795de79 : SUCCESS in 50m 33s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/bbf130c750aa48998e77ebe114990327 : SUCCESS in 57m 25s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/3825153bf03444d2bf63a5e48d6d11d2 : SUCCESS in 1h 00m 54s","accounts_in_message":[],"_revision_number":2},{"id":"3b254e3f066f8012b07b8998ae3ea55c478ef39d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-10 10:43:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"bd61e731679d62c30d474e9e40150fe4507495e2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-10 10:45:34.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/49451cf3e5dc4753b37a292b118e44cd\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d59178a099354ceda07a9cf5aebf7691 : SUCCESS in 49s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7203f49112c946a2985fd580155fbc8b : SUCCESS in 54s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f2cc2a1036bec02fa6119f870b9b56345afaaa0a","revisions":{"d8cb2062612e3f235799aeb5a11647a7c2ff1e4f":{"kind":"REWORK","_number":1,"created":"2024-11-26 14:47:23.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/68/936268/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/68/936268/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/68/936268/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/68/936268/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/68/936268/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/68/936268/1"}}},"commit":{"parents":[{"commit":"6a529bbc40e17401f9aaa92dca60532dfede90aa","subject":"Drop logic for absense of segregation rule","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/6a529bbc40e17401f9aaa92dca60532dfede90aa"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-26 14:44:28.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-26 14:46:24.000000000","tz":540},"subject":"Add policy rule for get quotas","message":"Add policy rule for get quotas\n\nThis introduces the missing policy customization capability for get\nquotas API, so that policy rules for all quota APIs can be customized.\n\nAlso fix missing target project_id in policy evaluation.\n\nChange-Id: I0e9a12670b8df448bed97448f8de9e3bbf207364\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/d8cb2062612e3f235799aeb5a11647a7c2ff1e4f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/d8cb2062612e3f235799aeb5a11647a7c2ff1e4f"}]},"branch":"refs/heads/master"},"f2cc2a1036bec02fa6119f870b9b56345afaaa0a":{"kind":"NO_CHANGE","_number":2,"created":"2024-11-26 15:34:00.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/68/936268/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/68/936268/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/68/936268/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/68/936268/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/68/936268/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/68/936268/2"}}},"commit":{"parents":[{"commit":"fac8aa64a0468f0c1bba3ad8e0a45623e7ad48db","subject":"Drop logic for absence of segregation rule","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/fac8aa64a0468f0c1bba3ad8e0a45623e7ad48db"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-26 14:44:28.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-26 15:34:00.000000000","tz":0},"subject":"Add policy rule for get quotas","message":"Add policy rule for get quotas\n\nThis introduces the missing policy customization capability for get\nquotas API, so that policy rules for all quota APIs can be customized.\n\nAlso fix missing target project_id in policy evaluation.\n\nChange-Id: I0e9a12670b8df448bed97448f8de9e3bbf207364\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/f2cc2a1036bec02fa6119f870b9b56345afaaa0a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/f2cc2a1036bec02fa6119f870b9b56345afaaa0a"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
