)]}'
{"id":"openstack%2Faodh~944129","triplet_id":"openstack%2Faodh~master~I502f9373d2572d9b7819fd779ff38b7562e2df31","project":"openstack/aodh","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-03-17 18:12:42.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I502f9373d2572d9b7819fd779ff38b7562e2df31","subject":"Cap maximum evaluation period and granularity","status":"MERGED","created":"2025-03-12 15:58:56.000000000","updated":"2025-03-17 18:15:22.000000000","submitted":"2025-03-17 18:12:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"944129","meta_rev_id":"9772ccac08c81804a59772c105bff8e8c5f83372","_number":944129,"virtual_id_number":944129,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-03-17 18:12:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"all":[{"value":2,"date":"2025-03-13 07:59:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":2,"date":"2025-03-17 17:02:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"all":[{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":1,"date":"2025-03-17 17:02:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-12 16:59:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-03-13 07:59:39.000000000","updated_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"reviewer":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"state":"REVIEWER"},{"updated":"2025-03-17 17:02:40.000000000","updated_by":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"reviewer":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"state":"REVIEWER"}],"messages":[{"id":"f6228a3955a9a439a464d99f8541bc970b6b0c34","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-03-12 15:58:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8fc8ed672d7f571f9b87eca21e84cdb529c9cd00","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-12 16:59:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dc4c4814b6374543aeb5667ca5baec34\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1dbddfd607a44c71bcd7eb929341b360 : SUCCESS in 4m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4edc93c875804c708647fc0997a6b778 : SUCCESS in 3m 30s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6eddb8da77604502ad7c618f2e7c6dd8 : SUCCESS in 6m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73bd8d349710480d98b16af077485a12 : SUCCESS in 4m 37s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/e7c3bf6e268f4060af1577f982b66ef5 : SUCCESS in 31m 39s\n- aodh-tempest-plugin-mysql-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/85a065eff23f4586938b79b3fad0437f : SUCCESS in 30m 11s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/616b6d36c5994b578bcaccf8956aa055 : SUCCESS in 27m 58s (non-voting)\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/a11ed193e19c401f94bda6a834b5c1d2 : SUCCESS in 54m 01s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/8bc410b5f3bb454487e27ce4aab253b3 : SUCCESS in 35m 51s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/60cb3d1cca784f109734441948ac3771 : SUCCESS in 41m 13s","accounts_in_message":[],"_revision_number":1},{"id":"c7457088040959385231ecf15a1c3498d92f44bc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-03-13 01:05:32.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"71a54e12d8e27060e6b8dc731a29326bd1caeaed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-13 02:11:19.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ddbb1105ae0c4a79b3f1ec6c84a1c267\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c64f8171c5c54e0baf8b9146b7d74c94 : SUCCESS in 4m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1de2593ba9324c4daeec78dec624af00 : SUCCESS in 4m 38s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ed80cda4739a4b59b98c85acb307069b : SUCCESS in 5m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d0ca67e7cb24226acfe20b46d9ff490 : SUCCESS in 6m 57s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/c8a56430152a442d9fee6ee337e16fde : SUCCESS in 38m 58s\n- aodh-tempest-plugin-mysql-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a611aab0d8a94472be0eff4aacd8704f : SUCCESS in 42m 45s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/f8f68b02719b4ae2b6ce0b68e0f3fa56 : SUCCESS in 33m 54s (non-voting)\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/db8862366d47432da0d4e8b128139761 : SUCCESS in 35m 10s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/a8948a37abf64932b016e951fa3cbf85 : SUCCESS in 52m 51s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/7611ba36ad9043368e0546690cf99367 : SUCCESS in 59m 18s","accounts_in_message":[],"_revision_number":2},{"id":"0ab8f2e0e68544b7c19791c756dbd9d86884d625","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2025-03-13 07:59:39.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"5097ea9f62c6178fb897ece44c6dbb8baf448ca9","author":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"date":"2025-03-17 17:02:40.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"e4c3179e410cb9787fb1aba1808ea80820526b78","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-17 17:03:22.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"80f32306ff999fc2e7f9c7e01e3010dfe6349ee7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-17 18:12:41.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0e394f34a47a45df89426ddc93dcb262\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b17974c75e5423ca187145bd28db459 : SUCCESS in 4m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0f8fe5f668364006952085c0fac9bb1c : SUCCESS in 7m 28s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/63f2bee5d6b8402fb7983dc39d57af23 : SUCCESS in 5m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6699dd63ff8e46d2bb667be7fa164482 : SUCCESS in 9m 12s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/85fb275d460545dcb22ccea300088582 : SUCCESS in 23m 53s\n- aodh-tempest-plugin-mysql-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3f53510581b74da3a52e4a4781a58d22 : SUCCESS in 34m 02s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/3667efac68c44b239c68c6c3f333c9d6 : SUCCESS in 28m 26s (non-voting)\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/981b84cc6bc04ab4aaba3828b7a3bf9d : SUCCESS in 49m 01s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/09def4bee0e949f199d716f4fd5e19e9 : SUCCESS in 33m 08s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/03d78c5d61de4ae0acb07321a9fe5a02 : SUCCESS in 42m 58s","accounts_in_message":[],"_revision_number":2},{"id":"1e75822c4943826998dc929819cbb408a0072e53","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-17 18:12:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"9772ccac08c81804a59772c105bff8e8c5f83372","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-17 18:15:22.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/50f5c8d96e5442a289b3ff7c4f829ef4\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d7833e2c81b4aeb988c941cf4b3846b : SUCCESS in 40s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7b369efc1c1a864a572d0ca8c6bf53562ae7864f","revisions":{"1e13b1fb3f21f76fbad9a48b922cba724ee0ccab":{"kind":"REWORK","_number":1,"created":"2025-03-12 15:58:56.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/29/944129/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/29/944129/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/29/944129/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/29/944129/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/29/944129/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/29/944129/1"}}},"commit":{"parents":[{"commit":"60ef9dd1b95d88d68d4f2f8d66fdb051b439cee6","subject":"Merge \"Fix missing nullable constraint\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/60ef9dd1b95d88d68d4f2f8d66fdb051b439cee6"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-03-12 15:45:58.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-03-12 15:58:14.000000000","tz":540},"subject":"Cap maximum evaluation period and granularity","message":"Cap maximum evaluation period and granularity\n\nSet 1 year as the maximum value of evaluation period and granularity to\navoid overflow errors caused by too long integers.\n\nCloses-Bug: #1315819\nChange-Id: I502f9373d2572d9b7819fd779ff38b7562e2df31\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/1e13b1fb3f21f76fbad9a48b922cba724ee0ccab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/1e13b1fb3f21f76fbad9a48b922cba724ee0ccab"}]},"branch":"refs/heads/master"},"7b369efc1c1a864a572d0ca8c6bf53562ae7864f":{"kind":"REWORK","_number":2,"created":"2025-03-13 01:05:32.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/29/944129/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/29/944129/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/29/944129/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/29/944129/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/29/944129/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/29/944129/2"}}},"commit":{"parents":[{"commit":"60ef9dd1b95d88d68d4f2f8d66fdb051b439cee6","subject":"Merge \"Fix missing nullable constraint\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/60ef9dd1b95d88d68d4f2f8d66fdb051b439cee6"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-03-12 15:45:58.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-03-13 01:05:13.000000000","tz":540},"subject":"Cap maximum evaluation period and granularity","message":"Cap maximum evaluation period and granularity\n\nto avoid overflow errors caused by too long integers.\n\nCloses-Bug: #1315819\nChange-Id: I502f9373d2572d9b7819fd779ff38b7562e2df31\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/7b369efc1c1a864a572d0ca8c6bf53562ae7864f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/7b369efc1c1a864a572d0ca8c6bf53562ae7864f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
