)]}'
{"id":"openstack%2Faodh~948971","triplet_id":"openstack%2Faodh~master~I07ac70065b276aa70eea5d5f1a5a0ac1a94f1242","project":"openstack/aodh","branch":"master","topic":"pip-23.1-support","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-06-05 09:00:03.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I07ac70065b276aa70eea5d5f1a5a0ac1a94f1242","subject":"add pyproject.toml to support pip 23.1","status":"MERGED","created":"2025-05-07 04:20:54.000000000","updated":"2025-06-05 09:01:10.000000000","submitted":"2025-06-05 09:00:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"948971-pip-23.1-support","meta_rev_id":"11949bcdb81cbdfb801ed53b745d0934a4148fd6","_number":948971,"virtual_id_number":948971,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-06-05 09:00:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"all":[{"value":2,"date":"2025-05-23 07:54:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-06-05 08:12:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"all":[{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-06-05 08:12:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-07 05:09:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-07 05:49:40.000000000","updated_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"reviewer":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"state":"CC"},{"updated":"2025-05-07 05:49:45.000000000","updated_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"reviewer":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"state":"REVIEWER"},{"updated":"2025-05-07 06:28:41.000000000","updated_by":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"reviewer":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"state":"REVIEWER"}],"messages":[{"id":"9ffc1ce4fe40bc36a244311693af6111e8e3b6d3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-05-07 04:20:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"100fc558427ed9691d9d80d476827c51f1797410","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-07 05:09:08.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/62c8aec890a74cfaa9d3e5a830db4b7a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb280bcd5d4641d29f54c2a4bac7962f : SUCCESS in 2m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/529843d679d74c649ce6179da9d13032 : SUCCESS in 3m 04s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f78cc893b2884ad6b8b30a7cbb3c36fe : SUCCESS in 4m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7691178b1f2549c597c3660f938b7da9 : SUCCESS in 6m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b32faa8b0b154deeba2dcbd46940fbee : SUCCESS in 7m 20s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/6ca1d2b32df74239942cdaa0795c6d76 : SUCCESS in 27m 20s\n- aodh-tempest-plugin-mysql-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4a1efd9a5dd1448b9762e9f5587000c9 : SUCCESS in 24m 25s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/59f309f9209d45d9a98fbcf2de1c4184 : SUCCESS in 30m 41s (non-voting)\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/bd9cd763f84b4527b867350552721d0d : SUCCESS in 42m 30s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/36d4f6e5afc242feb83cd7c0a9a74d6d : FAILURE in 10m 52s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/c32cd07ea4e34839af3021d57509db07 : FAILURE in 13m 16s","accounts_in_message":[],"_revision_number":1},{"id":"57d3951ca2b167ef126a04de621e4136eca2d9c2","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2025-05-07 05:49:40.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e1bcc166f91106c5a43d086e8a6ac54f9dd665fc","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2025-05-07 05:49:45.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"36bb3cf537c2f895458462060381c647aa798f29","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2025-05-07 05:50:11.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e9f911b8d06134cff3a914893b8b12b1bbf4838a","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-05-07 06:04:13.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f2789ced727d13fdc1c1afdb13731d5581fc54d8","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2025-05-07 06:28:41.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"ba8fc9cfe99030a8e52768194531b7fb9c279cdb","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-05-07 06:33:20.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ab7442640d977a4165bc386f9285b583f7b188ef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-05-12 10:21:35.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"987337dfd902d747309cd05e5996ab479edee0b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-12 10:53:59.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/892c3be690c644cd86fad166ae5cd99c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c39a8ce8682d4aa8bca8672b1b19bf8f : SUCCESS in 2m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6d6bfd83c44c46199922d89a7c3fd0ec : SUCCESS in 3m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/57e73a86080342f88f32c619e37d8a20 : SUCCESS in 4m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0097fc2a2a6c4b248accb6cc3b71d10e : SUCCESS in 4m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/162eb9e3128e4647aad976632e2584b2 : SUCCESS in 6m 28s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/37616220b06f414588b34136bd084d44 : FAILURE in 17m 55s\n- aodh-tempest-plugin-mysql-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/03bd30d5b1f14798bfd395580231c4af : SUCCESS in 17m 24s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/121491fe1a2a435b95ba83d729223f03 : SUCCESS in 26m 57s (non-voting)\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/6579f294ec924ab1b8219e09e7d5fdbe : SUCCESS in 24m 19s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/af11c9ea1c2b4e4e9a01e8a0113a3d49 : FAILURE in 20m 56s\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/ac787c6f4efb47858e56a807f55be206 : FAILURE in 27m 05s","accounts_in_message":[],"_revision_number":2},{"id":"0bc135e3cac5de53de8c860a83cf36dc9ceb99c4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-05-23 07:54:59.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"44df8c7db9c5b5951ed0876f23b50b8ba1e7e223","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-23 08:37:07.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0544d6eaa0d34e619768a624382644f5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/866012c273f14e749a6a4b581b7c45d1 : SUCCESS in 4m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/89e5d522da824da78020156156feaaaa : SUCCESS in 4m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/81f2be17763e4c3382f11bfe82a00431 : SUCCESS in 5m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6c0d74278a134415b5ce284d1ccacc28 : SUCCESS in 7m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c26fe52cb5fc4518a811e32deb80d742 : SUCCESS in 8m 03s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/7c84d3d9b4a5429d947d7523364e8b35 : SUCCESS in 35m 37s\n- aodh-tempest-plugin-mysql-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/77d763a0027d45f39e3c4f83b47f97ba : SUCCESS in 31m 06s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/5c352494f0ef478c97830823cc4feaca : SUCCESS in 29m 48s (non-voting)\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/0fb2ccf72f5a48f8b0fb52d5cd38c520 : SUCCESS in 37m 32s","accounts_in_message":[],"_revision_number":3},{"id":"f69c9a0061e1474cb83d8bd91f4d949f8d91c12d","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2025-06-05 08:12:30.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"149d40edc1350cd1f2842a1c40818dd549e113f7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-05 08:12:53.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"3888e6509f847e112bc03fa8029da5d93a3eea5c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-05 09:00:03.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/40a6ec8c8f7e40b5af7cedc23d7d911b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/06668b9ed86d4b949b9b102bf08aa660 : SUCCESS in 3m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/df50d8da8af243ebad9c0d52efbb7de1 : SUCCESS in 3m 17s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8d686f9b8fd5416cbf5c3cd8d7a6eb88 : SUCCESS in 4m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b88781e42fc429ca459f52a812d3e38 : SUCCESS in 6m 47s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/b6cdd715dd604fcc8500e8f42b41934a : SUCCESS in 15m 37s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/4b417b2c1a574836bb5dfd8f0fef2a24 : SUCCESS in 19m 04s (non-voting)\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/356cd0f71a7947fc8bf2f0b9af7df964 : SUCCESS in 42m 31s","accounts_in_message":[],"_revision_number":3},{"id":"a15b099b082392dd11af293fa7e344d22ebf27b2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-05 09:00:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"11949bcdb81cbdfb801ed53b745d0934a4148fd6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-05 09:01:10.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bb4c7bc3e59a4c2fbc95e104ff78562e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c76edd86688f434f941cc2ce5a976101 : SUCCESS in 52s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"538521ddf3afad6eee0c50430c8740e9c373f89e","revisions":{"78ac82067631d91b88572e0575a7a270f1639322":{"kind":"REWORK","_number":1,"created":"2025-05-07 04:20:54.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/71/948971/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/71/948971/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/71/948971/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/71/948971/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/71/948971/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/71/948971/1"}}},"commit":{"parents":[{"commit":"e5c561a4c47ed20460b166b62f769d5bb1e59484","subject":"Add possibility to get metrics for all projects","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/e5c561a4c47ed20460b166b62f769d5bb1e59484"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-05-07 04:20:44.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-05-07 04:20:44.000000000","tz":540},"subject":"add pyproject.toml to support pip 23.1","message":"add pyproject.toml to support pip 23.1\n\npip 23.1 removed the \"setup.py install\" fallback for projects\nthat do not have pyproject.toml and now uses a pyproject.toml\nwhich is vendored in pip.\nTo address that, this change adds the minimal pyproject.toml\nto enable pbr to be properly used to build editable wheels.\n\nThis is required to support installing devstack on\ncentos stream 9 and related distros with GLOBAL_VENV\u003dTrue\nWithout this change the wsgi scripts are not generated in\neditable mode. i.e. pip install -e /opt/stack/keystone\n\nSee https://pip.pypa.io/en/stable/news/#v23-1\nand https://github.com/pypa/pip/issues/8368 for more\ndetails on the removal of the fallback support.\n\nsetuptools v64.0.0 is used to support editable installs\nvia its PEP-660 implmentation\nhttps://github.com/pypa/setuptools/pull/3488\n\nCo-Authored-By: Sean Mooney \u003cwork@seanmooney.info\u003e\nChange-Id: I07ac70065b276aa70eea5d5f1a5a0ac1a94f1242\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/78ac82067631d91b88572e0575a7a270f1639322"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/78ac82067631d91b88572e0575a7a270f1639322"}]},"branch":"refs/heads/master"},"e966cfdbfb9f407107a828c055a5d7ae479f320a":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-05-12 10:21:35.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/71/948971/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/71/948971/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/71/948971/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/71/948971/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/71/948971/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/71/948971/2"}}},"commit":{"parents":[{"commit":"e3259778346bd1f982d883770ee35dc3229db07c","subject":"Merge \"Remove TODO note for old versions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/e3259778346bd1f982d883770ee35dc3229db07c"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-05-07 04:20:44.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-05-12 10:21:35.000000000","tz":0},"subject":"add pyproject.toml to support pip 23.1","message":"add pyproject.toml to support pip 23.1\n\npip 23.1 removed the \"setup.py install\" fallback for projects\nthat do not have pyproject.toml and now uses a pyproject.toml\nwhich is vendored in pip.\nTo address that, this change adds the minimal pyproject.toml\nto enable pbr to be properly used to build editable wheels.\n\nThis is required to support installing devstack on\ncentos stream 9 and related distros with GLOBAL_VENV\u003dTrue\nWithout this change the wsgi scripts are not generated in\neditable mode. i.e. pip install -e /opt/stack/keystone\n\nSee https://pip.pypa.io/en/stable/news/#v23-1\nand https://github.com/pypa/pip/issues/8368 for more\ndetails on the removal of the fallback support.\n\nsetuptools v64.0.0 is used to support editable installs\nvia its PEP-660 implmentation\nhttps://github.com/pypa/setuptools/pull/3488\n\nCo-Authored-By: Sean Mooney \u003cwork@seanmooney.info\u003e\nChange-Id: I07ac70065b276aa70eea5d5f1a5a0ac1a94f1242\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/e966cfdbfb9f407107a828c055a5d7ae479f320a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/e966cfdbfb9f407107a828c055a5d7ae479f320a"}]},"branch":"refs/heads/master","description":"Rebase"},"538521ddf3afad6eee0c50430c8740e9c373f89e":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2025-05-23 07:54:59.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/71/948971/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/71/948971/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/71/948971/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/71/948971/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/71/948971/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/71/948971/3"}}},"commit":{"parents":[{"commit":"2d57b2312ce6f7c0daeb08e3a14730f8bf5d23cd","subject":"Merge \"Remove CentOS Stream 9 jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/2d57b2312ce6f7c0daeb08e3a14730f8bf5d23cd"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-05-07 04:20:44.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-05-23 07:54:59.000000000","tz":0},"subject":"add pyproject.toml to support pip 23.1","message":"add pyproject.toml to support pip 23.1\n\npip 23.1 removed the \"setup.py install\" fallback for projects\nthat do not have pyproject.toml and now uses a pyproject.toml\nwhich is vendored in pip.\nTo address that, this change adds the minimal pyproject.toml\nto enable pbr to be properly used to build editable wheels.\n\nThis is required to support installing devstack on\ncentos stream 9 and related distros with GLOBAL_VENV\u003dTrue\nWithout this change the wsgi scripts are not generated in\neditable mode. i.e. pip install -e /opt/stack/keystone\n\nSee https://pip.pypa.io/en/stable/news/#v23-1\nand https://github.com/pypa/pip/issues/8368 for more\ndetails on the removal of the fallback support.\n\nsetuptools v64.0.0 is used to support editable installs\nvia its PEP-660 implmentation\nhttps://github.com/pypa/setuptools/pull/3488\n\nCo-Authored-By: Sean Mooney \u003cwork@seanmooney.info\u003e\nChange-Id: I07ac70065b276aa70eea5d5f1a5a0ac1a94f1242\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/538521ddf3afad6eee0c50430c8740e9c373f89e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/538521ddf3afad6eee0c50430c8740e9c373f89e"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
