)]}'
{"id":"openstack%2Faodh~949494","triplet_id":"openstack%2Faodh~master~Icf00b0f75532e0654915242daa64d397cde8dd4b","project":"openstack/aodh","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-05-15 11:46:53.000000000","reason":"Change was submitted"},"32968":{"account":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"last_update":"2025-05-15 11:13:43.000000000","reason":"\u003cGERRIT_ACCOUNT_32968\u003e replied on the change","reason_account":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"}}},"hashtags":[],"change_id":"Icf00b0f75532e0654915242daa64d397cde8dd4b","subject":"Remove CentOS Stream 9 jobs","status":"MERGED","created":"2025-05-12 12:01:31.000000000","updated":"2025-05-15 11:48:11.000000000","submitted":"2025-05-15 11:46:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"949494","meta_rev_id":"dfc159fcdf30b6427b177cea7d0edc7671d6ac1e","_number":949494,"virtual_id_number":949494,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-15 11:46:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"all":[{"value":2,"date":"2025-05-15 05:41:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":2,"date":"2025-05-15 11:13:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"all":[{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":1,"date":"2025-05-15 11:13:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-12 13:23:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-15 05:41:26.000000000","updated_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"reviewer":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"state":"REVIEWER"},{"updated":"2025-05-15 11:11:24.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"state":"REVIEWER"}],"messages":[{"id":"848ab5ca1e90002a79204222f0f23d4e24fb79fe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-05-12 12:01:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"216960f5518fe2e2847657e938f9b8c2091285b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-12 13:23:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fe23586363354b2892de5824cdf99e67\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/197faff34c244e0699e5e639ded60561 : SUCCESS in 3m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9a90b92dc335431fae40dbfd9cd60882 : SUCCESS in 4m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9accef613753470bb2128595cec52753 : SUCCESS in 5m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3a3b12aeab404d129896be854cf08042 : SUCCESS in 7m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ec4a00baa6c4e53ab64a4696f7a5c1f : SUCCESS in 8m 16s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/424f86c8fc0a47e3ae6bd8797c1a7e32 : SUCCESS in 37m 30s\n- aodh-tempest-plugin-mysql-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fbcb76f2a707432f89a4e3cc90cd752c : SUCCESS in 24m 39s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/ab1917175aa440118cdc982f77b19471 : SUCCESS in 14m 51s (non-voting)\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/fdf85250fecc4c069498a4e1a0dd4d3d : SUCCESS in 1h 15m 00s","accounts_in_message":[],"_revision_number":1},{"id":"daf6669f85bfd1782c761cb23320ea2fd499e84f","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2025-05-15 05:41:26.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1ea6a9fba79930d264bb759137eaa47281f8ecd3","author":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"date":"2025-05-15 11:13:43.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0a944b5cba4545d09ef6bee1b298dfd968e96566","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-15 11:14:15.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"26f486727500e5d478d6ca970bdcef395a5b2c0c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-15 11:46:53.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9f0a17a3d7584932a325f7e1e5106228\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f9935a7de15e4c40ad5ee836e5168fb2 : SUCCESS in 4m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9f7efe829ad04e17af1141ad0947c6f0 : SUCCESS in 4m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8e93da1e588d4cc5b601c89b85771567 : SUCCESS in 5m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bec7ccf6ea95455a8e29ddb1df8ba390 : SUCCESS in 7m 50s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/eaa4a946a9fa4eb0b76576628d8646bb : SUCCESS in 29m 50s\n- aodh-tempest-plugin-mysql-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ae8fcc7aefa242e4b0bfc043aa94e86d : SUCCESS in 14m 55s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/915b4afb4b6240148cb64396bad6f9c1 : SUCCESS in 27m 35s (non-voting)\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/6734e283f4944e32a33dc9d13fd40e33 : SUCCESS in 21m 02s","accounts_in_message":[],"_revision_number":1},{"id":"0dfc7c1126a9c2140d4cf1169e0099357a9cfdb3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-15 11:46:53.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"dfc159fcdf30b6427b177cea7d0edc7671d6ac1e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-15 11:48:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eaa4695c230c476a807322ac2aa4442b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15702c1e03774d92a8803bea0b722b57 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3b6ef16b68adc711dcfca14e3c46b0d3ad012c8b","revisions":{"3b6ef16b68adc711dcfca14e3c46b0d3ad012c8b":{"kind":"REWORK","_number":1,"created":"2025-05-12 12:01:31.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/94/949494/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/94/949494/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/94/949494/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/94/949494/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/94/949494/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/94/949494/1"}}},"commit":{"parents":[{"commit":"e5c561a4c47ed20460b166b62f769d5bb1e59484","subject":"Add possibility to get metrics for all projects","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/e5c561a4c47ed20460b166b62f769d5bb1e59484"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-05-12 12:00:47.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-05-12 12:01:18.000000000","tz":540},"subject":"Remove CentOS Stream 9 jobs","message":"Remove CentOS Stream 9 jobs\n\nThese no longer pass due to nova which removed Python 3.9 support.\nWe can add these jobs back once these jobs are migrated to CentOS\nStream 10.\n\nChange-Id: Icf00b0f75532e0654915242daa64d397cde8dd4b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/3b6ef16b68adc711dcfca14e3c46b0d3ad012c8b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/3b6ef16b68adc711dcfca14e3c46b0d3ad012c8b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
