)]}'
{"id":"openstack%2Faodh~950933","triplet_id":"openstack%2Faodh~master~I7fe406893282e763c242c567321b8272527b2687","project":"openstack/aodh","branch":"master","topic":"eventlet-removal","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-07-01 15:37:31.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I7fe406893282e763c242c567321b8272527b2687","subject":"Drop explicit executor argument","status":"MERGED","created":"2025-05-27 05:28:23.000000000","updated":"2025-07-01 15:39:19.000000000","submitted":"2025-07-01 15:37:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"950933-eventlet-removal","meta_rev_id":"fdf02bcf13c948a0323a3c751e8f93fcb7a35202","_number":950933,"virtual_id_number":950933,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-07-01 15:37:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"all":[{"value":2,"date":"2025-07-01 14:52:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-06-09 07:13:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"all":[{"value":1,"date":"2025-07-01 14:52:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-27 06:12:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-06-09 07:13:24.000000000","updated_by":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"reviewer":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"state":"REVIEWER"},{"updated":"2025-07-01 14:52:10.000000000","updated_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"reviewer":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"state":"REVIEWER"}],"messages":[{"id":"fe2cebd21459297ddd9ed425e2e7d67175f6db31","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-05-27 05:28:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2586272da48f8b9aab5ab2d1f5539ceaad65ae86","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-05-27 05:38:06.000000000","message":"Topic set to eventlet-removal","accounts_in_message":[],"_revision_number":1},{"id":"4bce827bab40a993672d2d74b47f1c95b4e9186d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-27 06:12:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f23121d313504e429b31b912bca66166\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea0579dd279d4c95899ff8e17e8bf230 : SUCCESS in 4m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bc0afc1f59b34592823988f4f402a36a : SUCCESS in 4m 23s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/83634f2656184ea6bdedd7840e8eb169 : SUCCESS in 4m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9b037556e49d42ae923647f215c944fc : SUCCESS in 6m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c1ca70204714ef19715f19e3766fca3 : SUCCESS in 9m 43s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/46f334f1cff04ac1aef7c6ed42c7b459 : SUCCESS in 34m 06s\n- aodh-tempest-plugin-mysql-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e724e11c1e124f48b24c6fa40ae72533 : SUCCESS in 31m 09s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/18397fb1287340f8a0048b2063c649ff : SUCCESS in 34m 02s (non-voting)\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/2d613fed5c864b65b9d9317d90bb44a2 : SUCCESS in 43m 18s","accounts_in_message":[],"_revision_number":1},{"id":"5d067d825f83309551f5c8ce8edf9a9c2f50a00d","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2025-06-09 07:13:24.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"0b14555ba6596969c86c43a86c028422067b5100","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2025-07-01 14:52:10.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"54744743056966dd55a74e357d7a24d689412b94","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2025-07-01 14:52:36.000000000","message":"Patch Set 1: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7bce480a0277a69ea376badf2aab50d38c478f3d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-01 14:53:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"37fa03fd47bdd9b7d9de5a48ebfdb973d12d7951","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-01 15:37:31.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/81f973f5b8fb4b5aacced71867bdf93e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8fe6dbc5f93349d0bccc2fac9eefab1e : SUCCESS in 3m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/20daecd531ba4063860761f5e8aba7ba : SUCCESS in 3m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/619c93fc544e405da514daf19be2ce4b : SUCCESS in 6m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0588d437120e49258d78f90c7dfa7f15 : SUCCESS in 4m 06s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/430ad76029664fe58d5b6a1111065736 : SUCCESS in 25m 46s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/46445e5a03d946d3983a9dc46df20e9e : FAILURE in 25m 14s (non-voting)\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/911efe49b1054d83b15b885bb48d3c30 : SUCCESS in 43m 09s","accounts_in_message":[],"_revision_number":1},{"id":"5cfc93f332a39c2cd393510efe96961a3898f0bd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-01 15:37:31.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"fdf02bcf13c948a0323a3c751e8f93fcb7a35202","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-01 15:39:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c69fd43cfe404d8aa07012e526c9c640\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6d3f2c1a1254fc2883589f2d6371aab : SUCCESS in 37s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7e03474a59796f3072011648ba50235ed0a8a510","revisions":{"7e03474a59796f3072011648ba50235ed0a8a510":{"kind":"REWORK","_number":1,"created":"2025-05-27 05:28:23.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/33/950933/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/33/950933/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/33/950933/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/33/950933/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/33/950933/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/33/950933/1"}}},"commit":{"parents":[{"commit":"2d57b2312ce6f7c0daeb08e3a14730f8bf5d23cd","subject":"Merge \"Remove CentOS Stream 9 jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/2d57b2312ce6f7c0daeb08e3a14730f8bf5d23cd"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-05-27 05:27:49.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-05-27 05:28:12.000000000","tz":540},"subject":"Drop explicit executor argument","message":"Drop explicit executor argument\n\nThe argument was deprecated in oslo.messaging 16.0.0, as part of huge\ndeprecation of eventlet support.\n\nIf the argument is not set, the appropriate executor is selected\naccording to whether eventlet monkey patching is done or not.\n\nChange-Id: I7fe406893282e763c242c567321b8272527b2687\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/7e03474a59796f3072011648ba50235ed0a8a510"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/7e03474a59796f3072011648ba50235ed0a8a510"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
