)]}'
{"id":"openstack%2Faodh~951305","triplet_id":"openstack%2Faodh~master~I32ddf41b7828b4c04c6dcfe17ba737954f775e45","project":"openstack/aodh","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-06-12 13:02:33.000000000","reason":"Change was submitted"},"4264":{"account":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"last_update":"2025-06-12 10:14:06.000000000","reason":"removed on reply"},"34975":{"account":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"last_update":"2025-06-12 12:17:37.000000000","reason":"\u003cGERRIT_ACCOUNT_34975\u003e replied on the change","reason_account":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"}}},"hashtags":[],"change_id":"I32ddf41b7828b4c04c6dcfe17ba737954f775e45","subject":"Drop extras for database backend","status":"MERGED","created":"2025-05-29 14:35:36.000000000","updated":"2025-06-12 13:04:17.000000000","submitted":"2025-06-12 13:02:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"951305","meta_rev_id":"74dd58f2020fe0ad113660d5c90e82bcab0f9ca4","_number":951305,"virtual_id_number":951305,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-06-12 13:02:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"all":[{"value":2,"date":"2025-06-12 10:14:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-06-12 12:17:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"all":[{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-06-12 12:17:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-29 14:59:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-06-02 06:27:52.000000000","updated_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"reviewer":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"state":"REVIEWER"},{"updated":"2025-06-05 06:42:34.000000000","updated_by":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"reviewer":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"state":"CC"},{"updated":"2025-06-12 12:17:37.000000000","updated_by":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"reviewer":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"state":"REVIEWER"}],"messages":[{"id":"f13bfe56017e2b7cff6109b10389f21feb8b92a1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-05-29 14:35:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c35d79eec31c84916311b8f0e39003f2221b8be4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-29 14:59:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/53ace85417e2404ebb2d061c7e860154\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/c4a68bbe677d4406a3720eb488afb1b9 : SUCCESS in 2m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ceb49870e05941a688c28b890b193f90 : SUCCESS in 3m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6c53ffda9bb341d79b94cfbf3cbba21a : SUCCESS in 4m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2a0cf6bf4fcb449dba28f0fcd0be1c1c : SUCCESS in 4m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/97f946ed966d4f33bf1c97062917d9bd : SUCCESS in 7m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91465180a13e4b97864b453e6d12b573 : SUCCESS in 8m 41s","accounts_in_message":[],"_revision_number":1},{"id":"849ccf8804100032c57b6fe49f12d10fe4f0bf65","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2025-06-02 06:27:52.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3fc9c906c94bbb3cc6cbffc2a65c8e75991dc980","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-06-03 13:37:51.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"1e6ef31ed7bdb9627082536be91bba21da1aaffb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-03 15:06:38.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/24f06230079f4fd3a30691ca710e4857\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/25d2f2e01c114593a8c3d4677c2c8fc1 : SUCCESS in 2m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8460a3896b744ed5b3aaf55590728e13 : SUCCESS in 4m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/62ed5f96595d4fc8aa50f3dffee54281 : SUCCESS in 4m 22s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/734f4031bf1644fc96a418c444025549 : SUCCESS in 4m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5c432ea22e314349b8e83ceecb473b9b : SUCCESS in 6m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9aa8d8ca9bfc4f71b0974604be544afb : SUCCESS in 7m 21s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/c69de114dedb4662b54c57d27c54c034 : SUCCESS in 35m 14s\n- aodh-tempest-plugin-mysql-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/eaa5fbf3c59b4b5e9cd29822f210f9c8 : SUCCESS in 31m 37s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/f45358f3e60a4b07b64022db723a5aca : SUCCESS in 37m 54s (non-voting)\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/79640b88e73a48fb8f51974ee120343b : SUCCESS in 44m 01s","accounts_in_message":[],"_revision_number":2},{"id":"1139d70541b6df6e9f6158026d408383140769ee","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2025-06-03 16:10:08.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"43a32bc132a2d08b39b87f99b705068ffc0a8a77","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2025-06-05 06:42:34.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4b2c664edae3ec1643b696fdad429c1ba6faed68","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-06-12 09:47:56.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"3afbb51eb3076580147a05a935f17c351f08dd4b","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-06-12 09:48:57.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"57d1cd349d57548d3cec86306702837c2e6f783c","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2025-06-12 10:14:06.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"6fab8a45d5fea4cbc54bad14b41d42360d155cbd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-12 10:20:13.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/399faff01d60429eafbf292b1ee1fcb2\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/228631bd6f5b4bce9b4d7e30e70cd235 : SUCCESS in 5m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8623c7356c484f49a07ab8097fab5b06 : SUCCESS in 6m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6676c8feebb5467f8f4c714953802ac6 : SUCCESS in 5m 51s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9b61e703919d4bbf9f6c154abb955418 : SUCCESS in 2m 34s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5124379cfc544d53998de468746f0377 : SUCCESS in 6m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11e37e53945a4eb89f6ea98d5615e304 : SUCCESS in 8m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2ac13b72e3bc45568f06a46120d4b293 : SUCCESS in 2m 53s\n- requirements-check https://zuul.opendev.org/t/openstack/build/6350d945a8814a6bad13a1b85855c3b4 : SUCCESS in 2m 46s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/f1c913477f4e4b6995b7ce98bb65ea86 : SUCCESS in 15m 04s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/53560aaf85704612888abfabf96004b9 : SUCCESS in 31m 18s (non-voting)\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/1ee10d4ef4114796867caaacb2a53b49 : SUCCESS in 20m 12s","accounts_in_message":[],"_revision_number":3},{"id":"6379b789a0f5cc1ce2b678338f56f08ead32f1a8","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2025-06-12 12:17:37.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"e0bb2e243a772ee5f0fcb55ce66dbf2f24d70058","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-12 12:18:01.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"eb91b34039eb359e9837e2aa5d6086cd60830e0e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-12 13:02:33.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fb59041676204a6baf10ece056a8229f\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/b29d0d44ed694e648735239f2ecc7a24 : SUCCESS in 4m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/682a7d228058484181de0ffb7c13e2d7 : SUCCESS in 1m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/12cdca25a83f40998585b41a266daf5c : SUCCESS in 5m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/122845f86c4e4c91adb14ed54a1c5ac5 : SUCCESS in 4m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ebb7593ea5bb465980e77ca2cd667c6e : SUCCESS in 3m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/22857f55335a48eca661f4a44d3d5b04 : SUCCESS in 4m 49s\n- requirements-check https://zuul.opendev.org/t/openstack/build/34e9a973961c4773a4eadb56d5f780e2 : SUCCESS in 2m 34s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/9c32adcebfd641a7a3dee7961dc8f465 : SUCCESS in 20m 18s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/993c668ed80f4e56945257b329f98f3b : SUCCESS in 26m 19s (non-voting)\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/a82a9cda8e1442ee9c7f033fe80cd127 : SUCCESS in 40m 19s","accounts_in_message":[],"_revision_number":3},{"id":"dcbe1d40b55dfc8919225432cbcfd4e5491da372","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-12 13:02:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"74dd58f2020fe0ad113660d5c90e82bcab0f9ca4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-12 13:04:17.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cbe773efdea94afb92e6c94d42b4ff5e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3989831aadc240a6b78d84b9707e1bcd : SUCCESS in 42s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4866baf69ffa48d493190dc2220ae922 : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"9049665c27aa2a13969e160ddd4ec553201443cd","revisions":{"a022ce8088fb9d109026888a1334e1a4898a8c78":{"kind":"REWORK","_number":1,"created":"2025-05-29 14:35:36.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/05/951305/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/05/951305/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/05/951305/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/05/951305/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/05/951305/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/05/951305/1"}}},"commit":{"parents":[{"commit":"2d57b2312ce6f7c0daeb08e3a14730f8bf5d23cd","subject":"Merge \"Remove CentOS Stream 9 jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/2d57b2312ce6f7c0daeb08e3a14730f8bf5d23cd"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-05-29 14:34:26.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-05-29 14:35:02.000000000","tz":540},"subject":"Drop extras for database backend","message":"Drop extras for database backend\n\nThese are maintained in oslo.db .\n\nAlso aodh itself does not directly depend on SQLAlchemy-Utils .\n\nChange-Id: I32ddf41b7828b4c04c6dcfe17ba737954f775e45\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/a022ce8088fb9d109026888a1334e1a4898a8c78"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/a022ce8088fb9d109026888a1334e1a4898a8c78"}]},"branch":"refs/heads/master"},"d3a6d1aba2e9569f7a629866e9155f21cc441a25":{"kind":"REWORK","_number":2,"created":"2025-06-03 13:37:51.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/05/951305/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/05/951305/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/05/951305/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/05/951305/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/05/951305/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/05/951305/2"}}},"commit":{"parents":[{"commit":"2d57b2312ce6f7c0daeb08e3a14730f8bf5d23cd","subject":"Merge \"Remove CentOS Stream 9 jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/2d57b2312ce6f7c0daeb08e3a14730f8bf5d23cd"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-05-29 14:34:26.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-06-03 13:37:27.000000000","tz":540},"subject":"Drop extras for database backend","message":"Drop extras for database backend\n\nThese are maintained in oslo.db .\n\nAlso aodh itself does not directly depend on SQLAlchemy-Utils .\n\nChange-Id: I32ddf41b7828b4c04c6dcfe17ba737954f775e45\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/d3a6d1aba2e9569f7a629866e9155f21cc441a25"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/d3a6d1aba2e9569f7a629866e9155f21cc441a25"}]},"branch":"refs/heads/master"},"9049665c27aa2a13969e160ddd4ec553201443cd":{"kind":"REWORK","_number":3,"created":"2025-06-12 09:47:56.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/05/951305/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/05/951305/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/05/951305/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/05/951305/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/05/951305/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/05/951305/3"}}},"commit":{"parents":[{"commit":"2d57b2312ce6f7c0daeb08e3a14730f8bf5d23cd","subject":"Merge \"Remove CentOS Stream 9 jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/2d57b2312ce6f7c0daeb08e3a14730f8bf5d23cd"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-05-29 14:34:26.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-06-12 09:47:42.000000000","tz":540},"subject":"Drop extras for database backend","message":"Drop extras for database backend\n\nThese are maintained in oslo.db .\n\nAlso aodh itself does not directly depend on SQLAlchemy-Utils .\n\nChange-Id: I32ddf41b7828b4c04c6dcfe17ba737954f775e45\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/9049665c27aa2a13969e160ddd4ec553201443cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/9049665c27aa2a13969e160ddd4ec553201443cd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
