)]}'
{"id":"openstack%2Faodh~954754","triplet_id":"openstack%2Faodh~master~I0fc49845af85fc716f3fb6221d81b047a18474cb","project":"openstack/aodh","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-02-18 17:16:10.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"I0fc49845af85fc716f3fb6221d81b047a18474cb","subject":"devstack: Use common function to manage keystoneauth options","status":"MERGED","created":"2025-07-11 15:42:51.000000000","updated":"2026-02-18 19:46:01.000000000","submitted":"2026-02-18 19:45:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"954754","meta_rev_id":"90b1f9ac98d633991dbf8898575df27524f2ecbf","_number":954754,"virtual_id_number":954754,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-18 19:45:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"all":[{"value":2,"date":"2025-07-16 11:00:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":2,"date":"2025-10-15 10:55:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"all":[{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":1,"date":"2025-10-15 10:55:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-07-11 16:27:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-07-16 11:00:32.000000000","updated_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"reviewer":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"state":"REVIEWER"},{"updated":"2025-10-15 10:55:58.000000000","updated_by":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"reviewer":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"state":"REVIEWER"}],"messages":[{"id":"ff2495277a83635f9d55a43ce0a96f1b8182b196","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-07-11 15:42:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d3b5a21fba9a869b9918bab7212b6c4f7381a0c7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-11 16:27:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/084c7d630f9543bc8a798ebd190f9df7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/708d061396ef403a8ba03abae31b6e03 : SUCCESS in 3m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bd962da787f64506ae5ab9f5899703ed : SUCCESS in 4m 23s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/022926bf8b6f43bbb7fa621474d353cb : SUCCESS in 4m 56s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5d0ce70c85364ad3aa760d4c3547767b : SUCCESS in 4m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1948b635dbdc43f0ae5213462c75436f : SUCCESS in 8m 22s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/40356516231e437aa5d14d18b4b1db59 : SUCCESS in 32m 39s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/a06f0d7b85004e9fa0f423850880c25b : FAILURE in 27m 01s (non-voting)\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/b6ed6e828e974c61b47af728032d272c : SUCCESS in 42m 57s","accounts_in_message":[],"_revision_number":1},{"id":"2901c4d0f2f440d861d5284a1e02699ea664464d","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2025-07-16 11:00:32.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"dc1229fe681dd781b0b75a4b6abbc1f9b218eac0","author":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"date":"2025-10-15 10:55:58.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fa93ec58ac02996c4b64c8776edccfae8191a449","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-15 10:56:02.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/381ed5bddd314ca8a8079cd3e2c05378\n\n\nWarning:\n  Change 954751,5 in project openstack/devstack does not share a change queue with 954754 in project openstack/aodh","accounts_in_message":[],"_revision_number":1},{"id":"4bf7f98765fb0a49808688f69c07d3db1f478a66","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-02-18 17:16:10.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1fb998cef8a0bf70480ea349b41e561b42e95a9f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-18 18:28:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4fff2bc3f163482db89b151c7fca3c34\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2aa9fc0d77594de3bcb7c636c02c8302 : SUCCESS in 3m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7f51a01f25bf465f84fe1dc68cb921e8 : SUCCESS in 4m 36s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8d7f8006076e41779bb09769b4370912 : SUCCESS in 2m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9ee90dabe594575acea13c7b09d61d3 : SUCCESS in 4m 59s\n- grenade-aodh https://zuul.opendev.org/t/openstack/build/d858b47999fd4095b13916fb7bb468c1 : SUCCESS in 1h 06m 33s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/40f3498b94f942cf93a12bc0fedf7bd8 : SUCCESS in 30m 42s","accounts_in_message":[],"_revision_number":1},{"id":"5c43293be72c07d6da72975c75e6a9f5b3bdebbf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-18 18:29:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"945114c56fab56ade195e750583e3f539ee182b3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-18 19:45:11.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8d1a5e44670242f3aa70e488c7ba3b09\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4fcc7a81d46044979dcf0f907e76438f : SUCCESS in 2m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b03d41af680c49a1a4fa18b5ab2eb2ab : SUCCESS in 4m 30s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ee466d4080be42c183e21be7b23a68d2 : SUCCESS in 4m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb0f05ac3ae947f08b7aba69717e4e64 : SUCCESS in 4m 49s\n- grenade-aodh https://zuul.opendev.org/t/openstack/build/9cbdcebd22f44b39b4bcab5d333e710e : SUCCESS in 1h 11m 14s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/aa3f33ed4c284808a05736dd9078b843 : SUCCESS in 52m 22s","accounts_in_message":[],"_revision_number":1},{"id":"a59a39e1a4e2dc4889d6a5256529be78b305179f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-18 19:45:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"90b1f9ac98d633991dbf8898575df27524f2ecbf","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-18 19:46:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/48bb441779124d7d81ece7edcb0bd20b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad190f1118c446ee94d22065f36cfb38 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8da25aa80ff3f1ff2ea5c4dd9fe0f8e55c42b4c8","revisions":{"8da25aa80ff3f1ff2ea5c4dd9fe0f8e55c42b4c8":{"kind":"REWORK","_number":1,"created":"2025-07-11 15:42:51.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/54/954754/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/54/954754/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/54/954754/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/54/954754/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/54/954754/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/54/954754/1"}}},"commit":{"parents":[{"commit":"2a563b221e0ec964f646b0f7f31e28d163b6b1d7","subject":"sqlalchemy: Use built-in declarative","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/2a563b221e0ec964f646b0f7f31e28d163b6b1d7"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-07-11 15:42:27.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-07-11 15:42:42.000000000","tz":540},"subject":"devstack: Use common function to manage keystoneauth options","message":"devstack: Use common function to manage keystoneauth options\n\nDepends-on: https://review.opendev.org/954751\nChange-Id: I0fc49845af85fc716f3fb6221d81b047a18474cb\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/8da25aa80ff3f1ff2ea5c4dd9fe0f8e55c42b4c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/8da25aa80ff3f1ff2ea5c4dd9fe0f8e55c42b4c8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
