)]}'
{"id":"openstack%2Faodh~955417","triplet_id":"openstack%2Faodh~master~Idc0eecc8116dbcd171b3fbd058491eac46bf0a36","project":"openstack/aodh","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-08-12 16:13:15.000000000","reason":"Change was submitted"},"32968":{"account":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"last_update":"2025-08-12 16:13:15.000000000","reason":"Change was submitted"},"4264":{"account":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"last_update":"2025-08-12 14:54:39.000000000","reason":"\u003cGERRIT_ACCOUNT_4264\u003e replied on the change","reason_account":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"}},"34975":{"account":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"last_update":"2025-08-04 10:00:44.000000000","reason":"\u003cGERRIT_ACCOUNT_34975\u003e replied on the change","reason_account":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"}}},"hashtags":[],"change_id":"Idc0eecc8116dbcd171b3fbd058491eac46bf0a36","subject":"Add cascade to delete related rows automatically","status":"MERGED","created":"2025-07-19 03:21:22.000000000","updated":"2025-08-12 16:14:31.000000000","submitted":"2025-08-12 16:13:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"955417","meta_rev_id":"043a47797c016599f15af085134d7eed86aa61b5","_number":955417,"virtual_id_number":955417,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-08-12 16:13:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"all":[{"value":2,"date":"2025-08-12 14:54:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-08-04 10:00:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"all":[{"value":1,"date":"2025-08-12 14:54:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"}],"reviewers":{"REVIEWER":[{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-07-19 04:10:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-08-04 08:42:29.000000000","updated_by":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"reviewer":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"state":"REVIEWER"},{"updated":"2025-08-12 12:45:02.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"state":"REVIEWER"},{"updated":"2025-08-12 12:45:02.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"state":"REVIEWER"}],"messages":[{"id":"ef06cdc4d8468dce12f5d9345cc9edbcdb829067","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-07-19 03:21:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7a7e73ac414231f67ec528319ac4d06f304b74ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-19 04:10:05.000000000","message":"Patch Set 1: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/664d4fa6648743a6868dc33d1a2289ca\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/445d64e809aa428082f470ccb5c6d2be : FAILURE in 5m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f88df68550534a43b6db8a73dc0c4a46 : SUCCESS in 2m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8b56bdf2c960476eb99b98b3281253fc : SUCCESS in 8m 20s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bf83033d99cb456e84f24bcca6fe4bad : SUCCESS in 10m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c0fb68a32374c88b15a7255d3144182 : SUCCESS in 4m 16s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/bcad01bd88104974ab2177ac86ea574a : SUCCESS in 15m 46s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/ab7c0f65c47c444c9768489c122d6f64 : FAILURE in 11m 40s (non-voting)\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/df3b54dbed6f44e4855aa0ff670b91ef : SUCCESS in 47m 01s","accounts_in_message":[],"_revision_number":1},{"id":"243a4ba2e7491764818300cfc4d44b37a22579d3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-07-19 07:55:59.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"5bf9ed13e116aee013c01adb95efc1eb3a1b5ed2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-19 08:43:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6e27ed681ea04c269f3b92665895a6e8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/043c9d0c3de84750aec68aa08ac9f5db : SUCCESS in 2m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3a861fb4ec4241b8b13013c8c5985f5c : SUCCESS in 4m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/633f74f156404097aefb82d911c0ba5d : SUCCESS in 5m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/92645e272ecd4a72bec978c321ac47d2 : SUCCESS in 9m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aaaf69e657ab4c56af954127a32b5e82 : SUCCESS in 5m 56s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/54bdb1fa2ae34f99a6ca578033a0d816 : SUCCESS in 27m 12s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/35d198c6abe2490e995704dad55b6d73 : FAILURE in 27m 58s (non-voting)\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/059434b26f8b47a0a31fc1881be33a4c : SUCCESS in 42m 10s","accounts_in_message":[],"_revision_number":2},{"id":"3bb55fb25416646f1a87f96787926791a73cb5cb","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2025-08-04 08:42:29.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1c148ffdca67bb1897c9aa20cbcd182fc42b77ce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-08-04 09:11:20.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"34c300a7ec95f537c97fa9d6a8ed587852fc2961","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-08-04 09:11:51.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"1651039d6592b9d48b8c2c6eac94e253010c6033","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-04 09:55:56.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c9472191894d4a4da4fff689066ba232\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac576ea7f3b4461a81d3bc78c30838bf : SUCCESS in 4m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/122f6df9195d43de9fec6c8c05ca27c4 : SUCCESS in 5m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/193bd6a09c87409187a5621448a99434 : SUCCESS in 5m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ede766e35fc44fe69cf4759a8190e4c9 : SUCCESS in 6m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5211503b82554bfca668329a5018ef12 : SUCCESS in 8m 09s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/25a7cf47aaeb4382a67d8ac86b640331 : SUCCESS in 40m 00s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/b77a672ca82141e595a5b316df581d3d : FAILURE in 24m 42s (non-voting)\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/7a07e9dd7e4349e49380d2d28969cfcd : SUCCESS in 40m 10s","accounts_in_message":[],"_revision_number":3},{"id":"fae1ea0702299bfb2324955b445f929ee5da91ab","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2025-08-04 10:00:44.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"551de11d65d294d4631337020460b48fe85d1ddd","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2025-08-12 14:54:39.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"dfd2b07cd021faf250b78b76b71e3100d9f307cf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-12 14:55:10.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"e4828ec16032eda20595d47420093a94205b564a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-12 16:13:15.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9ffa4d3ebb464d5dbd92ffb637cea65e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e738d94a192435fbe442cb50d45c381 : SUCCESS in 5m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3d7988b9d8b94f8f8bd8e70a189e1799 : SUCCESS in 4m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/79f3ea7cd03e47dbbcf8e32c707fb4ac : SUCCESS in 6m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b281c78a2f9a4e98aa17ddb5d7ff1251 : SUCCESS in 9m 10s\n- grenade-aodh https://zuul.opendev.org/t/openstack/build/d797eeaebee849819b4cbfe99ef4920f : SUCCESS in 1h 13m 07s (non-voting)\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/5d42f74f31b1417b913cf5986da413db : SUCCESS in 36m 27s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/ec4bbf3367b44bf3abfc79639202354d : FAILURE in 29m 06s (non-voting)\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/02c7e672c80c4d829ae2aae831bce115 : SUCCESS in 47m 34s","accounts_in_message":[],"_revision_number":3},{"id":"46634133ec01d2112fbe153af3298c65b0bb4947","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-12 16:13:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"043a47797c016599f15af085134d7eed86aa61b5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-12 16:14:31.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eaf968552d7244b5a8d36615e3ed19c8\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6cbcf97987a4813b8655959b7d0ab62 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"a72afc36d0691f6797552bd7b27c76e1753aba05","revisions":{"2b7eed919cc9b41449bddf393bab2cb5a908fc4d":{"kind":"REWORK","_number":1,"created":"2025-07-19 03:21:22.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/17/955417/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/17/955417/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/17/955417/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/17/955417/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/17/955417/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/17/955417/1"}}},"commit":{"parents":[{"commit":"f63d09384c28acfc5586c14f20205c71c58ade3b","subject":"Fix missing FK constraint from alarm history to alarm","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/f63d09384c28acfc5586c14f20205c71c58ade3b"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-07-19 03:14:42.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-07-19 03:21:06.000000000","tz":540},"subject":"Add cascade to delete related rows automatically","message":"Add cascade to delete related rows automatically\n\n... according to the remaining TODO comment.\n\nChange-Id: Idc0eecc8116dbcd171b3fbd058491eac46bf0a36\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/2b7eed919cc9b41449bddf393bab2cb5a908fc4d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/2b7eed919cc9b41449bddf393bab2cb5a908fc4d"}]},"branch":"refs/heads/master"},"7956aa4b544a4d4eed133968465c23ad7445ec7e":{"kind":"REWORK","_number":2,"created":"2025-07-19 07:55:59.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/17/955417/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/17/955417/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/17/955417/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/17/955417/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/17/955417/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/17/955417/2"}}},"commit":{"parents":[{"commit":"f63d09384c28acfc5586c14f20205c71c58ade3b","subject":"Fix missing FK constraint from alarm history to alarm","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/f63d09384c28acfc5586c14f20205c71c58ade3b"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-07-19 03:14:42.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-07-19 07:55:46.000000000","tz":540},"subject":"Add cascade to delete related rows automatically","message":"Add cascade to delete related rows automatically\n\n... according to the remaining TODO comment.\n\nChange-Id: Idc0eecc8116dbcd171b3fbd058491eac46bf0a36\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/7956aa4b544a4d4eed133968465c23ad7445ec7e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/7956aa4b544a4d4eed133968465c23ad7445ec7e"}]},"branch":"refs/heads/master"},"a72afc36d0691f6797552bd7b27c76e1753aba05":{"kind":"REWORK","_number":3,"created":"2025-08-04 09:11:20.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/17/955417/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/17/955417/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/17/955417/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/17/955417/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/17/955417/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/17/955417/3"}}},"commit":{"parents":[{"commit":"f63d09384c28acfc5586c14f20205c71c58ade3b","subject":"Fix missing FK constraint from alarm history to alarm","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/f63d09384c28acfc5586c14f20205c71c58ade3b"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-07-19 03:14:42.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-04 09:11:20.000000000","tz":0},"subject":"Add cascade to delete related rows automatically","message":"Add cascade to delete related rows automatically\n\n... according to the remaining TODO comment.\n\nChange-Id: Idc0eecc8116dbcd171b3fbd058491eac46bf0a36\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/a72afc36d0691f6797552bd7b27c76e1753aba05"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/a72afc36d0691f6797552bd7b27c76e1753aba05"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
