)]}'
{"id":"openstack%2Faodh~957030","triplet_id":"openstack%2Faodh~master~Ia71c34e5bf99be748d1ff9812b99149c7d29fff9","project":"openstack/aodh","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-08-27 08:06:31.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},"32968":{"account":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"last_update":"2025-08-27 07:12:18.000000000","reason":"\u003cGERRIT_ACCOUNT_32968\u003e replied on the change","reason_account":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"}},"34975":{"account":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"last_update":"2025-08-27 10:25:51.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia71c34e5bf99be748d1ff9812b99149c7d29fff9","subject":"Remove [DEFAULT] http_timeout","status":"MERGED","created":"2025-08-11 15:31:45.000000000","updated":"2025-08-27 10:29:18.000000000","submitted":"2025-08-27 10:25:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"957030","meta_rev_id":"05978644d5a0f5d7f369ff61275fef763c6e0eb3","_number":957030,"virtual_id_number":957030,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-08-27 10:25:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"all":[{"value":2,"date":"2025-08-12 14:57:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":2,"date":"2025-08-27 07:12:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"all":[{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":1,"date":"2025-08-27 07:12:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"}],"reviewers":{"REVIEWER":[{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-11 16:39:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-08-12 14:57:13.000000000","updated_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"reviewer":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"state":"REVIEWER"},{"updated":"2025-08-26 14:16:31.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"state":"REVIEWER"},{"updated":"2025-08-26 14:16:31.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"state":"REVIEWER"}],"messages":[{"id":"d55b0dee70af000d3b5055143863d9fed83b2763","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-08-11 15:31:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6c86feb8ad15f26ab39f8e92168e0bbb32290714","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-11 16:39:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ee45002b54994d259a7bdbdf8c95ee73\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/09dd9a26cee9420fb5587ef09071e5dc : SUCCESS in 2m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/79d21d9858ac417290cd34d58480208a : SUCCESS in 4m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/62da2d9ab775464b88437f45ee034c16 : SUCCESS in 5m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c621e77fb8d14bef9b62ab967c84c589 : SUCCESS in 3m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3700c63c412a41af8e719a03f4550aa9 : SUCCESS in 7m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/239396356be74a74969ec43294414bd9 : SUCCESS in 6m 54s\n- grenade-aodh https://zuul.opendev.org/t/openstack/build/c82bfbf38be0431b8ab1279772247f7b : SUCCESS in 1h 02m 24s (non-voting)\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/a73db01d1d774786a8af7f6a234b65bc : SUCCESS in 29m 02s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/85c48a33729547b5a8fe832c79578389 : FAILURE in 25m 51s (non-voting)\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/9968ac645f904efca91c44d52dc9e64e : SUCCESS in 34m 29s","accounts_in_message":[],"_revision_number":1},{"id":"79e5fd966a28b40cf7e4c0fbe039586e4357122c","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2025-08-12 14:57:13.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3ef7c306f37c0641e2e11449b69b8011c8ea3b90","author":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"date":"2025-08-27 07:12:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"52ff848531bfdc04c27c26267406c6b2fa02b2a0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-27 07:12:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"039717a1dfce996be673caa6a7d07fca2ce893ef","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-27 08:03:45.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2bdb92daa0534b678caec3758f79049f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b238abed56174062b082f0e3443071e6 : SUCCESS in 3m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3092fc4e364e43ee9a14c253f391397c : SUCCESS in 4m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/bb2ffba50caf484193e8e668f7cabe25 : SUCCESS in 3m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9032f3a40c02412f8d7b36d54cba5cf4 : SUCCESS in 8m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d116f245300f44f6a931b1db6ddc55bc : SUCCESS in 6m 38s\n- grenade-aodh https://zuul.opendev.org/t/openstack/build/18996062ce524c058202c06f72779ecc : FAILURE in 25m 46s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/57b4df65be80488289f6cf2de65dffd8 : SUCCESS in 45m 28s","accounts_in_message":[],"_revision_number":1},{"id":"31378a886a2c54fe2f6abe70c1f7f3911af4cdb4","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-08-27 08:06:31.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f2ce84e2d3f4ebfdbe5244eb14591724912ba4d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-27 09:13:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/90ab7a778e2045a2873e0a2a9f614bbd\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5cb8e1078414416a22496bc2998efa3 : SUCCESS in 5m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8b52be35e0ad47efae1dd429566d657a : SUCCESS in 4m 38s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ab4afd5d648546ac8bc4b294993b21d5 : SUCCESS in 4m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/62b51b0e343d47bf991d28aab0151009 : SUCCESS in 4m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d2e0f2bcfaa46838250d995223b1765 : SUCCESS in 4m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/321beb2db71d4f33bbaacd8b6df082c5 : SUCCESS in 3m 50s\n- grenade-aodh https://zuul.opendev.org/t/openstack/build/79b4d13b2eb044b5b1d34399da52e9b5 : SUCCESS in 1h 02m 21s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/b3c0dd8a307b4176a66733acd8cc4d8b : SUCCESS in 26m 21s","accounts_in_message":[],"_revision_number":1},{"id":"aa20becbab82bac9e7102ef39a193cc1b5735385","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-27 09:14:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"db366930d44d343bf45c85e324b6f62bae5b012c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-27 10:25:51.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ead61832137d4975a54b2ebc507bd186\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f97f7f118ed848cdab6ee510e37c7182 : SUCCESS in 4m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/242861ec843745678a2d893bff001dc7 : SUCCESS in 4m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e6d54b21ae38443384690c4c26a28e3a : SUCCESS in 5m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b50ea0b8140494794b27aec930958e2 : SUCCESS in 8m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/61ffe35ae278401a93c8a3dbaba6462c : SUCCESS in 6m 44s\n- grenade-aodh https://zuul.opendev.org/t/openstack/build/a8476e4960664a59b55e229ebc97f8d4 : SUCCESS in 1h 04m 40s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/4495f696f155461ea541d388614fe46c : SUCCESS in 26m 58s","accounts_in_message":[],"_revision_number":1},{"id":"8839b76eca9f39087a16478707646e8fdce2ad95","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-27 10:25:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"05978644d5a0f5d7f369ff61275fef763c6e0eb3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-27 10:29:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ef84a031df2e4d089220e03472f0fd89\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10b0f53b0efe4f5e8ea65a0c54ef8bd9 : SUCCESS in 1m 08s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5dfb7c4804df42d3b1a8931e5607d4d1 : SUCCESS in 3m 19s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3aecf348ee1981452c11a40d54ff82402f64358f","revisions":{"3aecf348ee1981452c11a40d54ff82402f64358f":{"kind":"REWORK","_number":1,"created":"2025-08-11 15:31:45.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/30/957030/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/30/957030/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/30/957030/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/30/957030/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/30/957030/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/30/957030/1"}}},"commit":{"parents":[{"commit":"b5125ead7588644401f0feb82cf69b37db86a59d","subject":"Merge \"Add grenade job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/b5125ead7588644401f0feb82cf69b37db86a59d"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-11 15:30:59.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-11 15:31:29.000000000","tz":540},"subject":"Remove [DEFAULT] http_timeout","message":"Remove [DEFAULT] http_timeout\n\nThis option was deprecated a few years ago[1].\n\n[1] 12ee78cdef1b3ca26ada16f05af795f5ad4f3da4\n\nChange-Id: Ia71c34e5bf99be748d1ff9812b99149c7d29fff9\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/3aecf348ee1981452c11a40d54ff82402f64358f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/3aecf348ee1981452c11a40d54ff82402f64358f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
