)]}'
{"id":"openstack%2Faodh~957279","triplet_id":"openstack%2Faodh~master~I6f1b0a1bc67376dca3e00585f04bec697781113c","project":"openstack/aodh","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-11-21 08:25:53.000000000","reason":"Change was submitted"},"4264":{"account":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"last_update":"2025-11-21 07:12:04.000000000","reason":"\u003cGERRIT_ACCOUNT_4264\u003e replied on the change","reason_account":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"}},"32968":{"account":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"last_update":"2025-10-15 11:03:37.000000000","reason":"\u003cGERRIT_ACCOUNT_32968\u003e replied on the change","reason_account":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"}},"34975":{"account":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"last_update":"2025-11-21 08:25:53.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6f1b0a1bc67376dca3e00585f04bec697781113c","subject":"Rbac restrictions to prometheus alarms","status":"MERGED","created":"2025-08-13 19:23:48.000000000","updated":"2025-11-21 08:27:01.000000000","submitted":"2025-11-21 08:25:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"957279","meta_rev_id":"d8a448ce4b046b4974841e9bb5d6373416d732c5","_number":957279,"virtual_id_number":957279,"owner":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-11-21 08:25:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"all":[{"value":2,"date":"2025-10-15 11:03:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},{"value":2,"date":"2025-11-21 07:12:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"all":[{"value":0,"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},{"value":1,"date":"2025-11-21 07:12:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}],"reviewers":{"REVIEWER":[{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-13 20:39:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-10-14 08:05:09.000000000","updated_by":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"reviewer":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"state":"REVIEWER"},{"updated":"2025-10-14 08:05:09.000000000","updated_by":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"reviewer":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"state":"REVIEWER"},{"updated":"2025-10-14 08:05:09.000000000","updated_by":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"}],"messages":[{"id":"f58ea1458c4374efdb0fcf8e3cb60be90b03f652","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2025-08-13 19:23:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"68750c1f51459af1e8ac5044b07f50ab26c7c0b6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2025-08-13 19:24:46.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"7aa706ccf646cb3f28acffbc79cba9dd03f34751","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-13 20:39:47.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9f22bb8f89ac4758bc1247a056cd1e1c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f3546f2c815c40ec8cd77cc523f0b097 : SUCCESS in 3m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/aaf3b6e31c694eb0a9c2147ba99bf2a2 : SUCCESS in 4m 03s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f39d788687944844bedeb731758d8d67 : SUCCESS in 4m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5c9ca76b2c704eaa8f6f9b90fcb3db91 : SUCCESS in 7m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8344fbc7d4ba466eaa5ab7fc88f8a6dd : SUCCESS in 9m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9164a41ec2724eafb6108d728c50f8e6 : SUCCESS in 5m 39s\n- requirements-check https://zuul.opendev.org/t/openstack/build/eefc222f87844fc69a2afe2a79342624 : SUCCESS in 3m 43s\n- grenade-aodh https://zuul.opendev.org/t/openstack/build/4197cec5c8db49b4af19af9123da056d : SUCCESS in 1h 09m 20s (non-voting)\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/ce6c9bcb592e44e3931e04bb60b5c07c : FAILURE in 57m 07s","accounts_in_message":[],"_revision_number":2},{"id":"307cd290258d02099d98f8cafd972ee71428a1cb","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2025-08-14 07:52:33.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":2},{"id":"ec0ee11660c93bfc11ef50f411ac2c481653180f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2025-08-15 07:43:40.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"816b0dbb9650912486468ef8f8c15684be2d9e3a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-15 08:44:57.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5599add6d88b4c45bd158e5f615b002b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cec677080f2d40eda27aee79baec4972 : SUCCESS in 4m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/736454716a26429ebd5741209e534145 : SUCCESS in 3m 20s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a90c9eccee5a411d915563b7e1d22023 : SUCCESS in 6m 38s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/88f0ebc27a794fc89be7bf7198fb11e5 : SUCCESS in 4m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17a15dc666cb4f17a389513c0beaaa92 : SUCCESS in 9m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/947b983160854f24951bc3be065ee663 : SUCCESS in 7m 10s\n- requirements-check https://zuul.opendev.org/t/openstack/build/0e73433624af48a69968109f46522538 : SUCCESS in 2m 31s\n- grenade-aodh https://zuul.opendev.org/t/openstack/build/a99226d537b647ec9db0c60415297081 : SUCCESS in 58m 21s (non-voting)\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/7ea55c83d258453dbfcf213f94e780ab : SUCCESS in 25m 52s","accounts_in_message":[],"_revision_number":3},{"id":"dcdb22145031ae3e5806858c8e42c080d1558deb","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2025-10-07 12:38:28.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"84c564239e06313b2df94075defba2f8e72ceb60","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-07 13:11:14.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6c5f89dad28c4a34b55da6462a5449b7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/15a88dd048864b2884df8d3a83c744a8 : SUCCESS in 3m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6de56bc0a89d405489a7570c29eca22a : SUCCESS in 4m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ea923303fa0e49e8bb424ce47fd68af2 : FAILURE in 5m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e8694c652c47464fae5a1731523e5356 : FAILURE in 8m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f612609663414fa895bd1578266a51b3 : SUCCESS in 4m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b5903daae70b40f7a9a61e63d779619e : SUCCESS in 6m 18s\n- requirements-check https://zuul.opendev.org/t/openstack/build/fb41f29b522d4826a31fc32ab60de6c5 : SUCCESS in 2m 50s\n- grenade-aodh https://zuul.opendev.org/t/openstack/build/8aeae39505cf4dd69fa4a81950d79df1 : SUCCESS in 31m 36s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/c33ba2e191f847c8ab068f4c9bc40ca2 : SUCCESS in 21m 41s","accounts_in_message":[],"_revision_number":4},{"id":"34b9f7926a0ad2080b4f0933c8b7440ff3c7768b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2025-10-07 13:21:27.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"2b450f75769ea77b4b929826e2430d5c6a0a289e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-07 14:35:04.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6092d5628113459db0e9ce43bac235ad\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc5695e4fdd84b6ebbbffb58c14a86ae : SUCCESS in 3m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/186455f6e8564e49ae185d9c79a763ff : SUCCESS in 4m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5b69d1bc86c64a0d9049917fa66d9464 : SUCCESS in 5m 43s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/90a7af175e0245c0b30eb0ed2fddf80d : SUCCESS in 9m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73e2a41c1a6e4c10934d22da7d0bd4f1 : SUCCESS in 9m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/76b8b070f97d42129c3e37992f50a43d : SUCCESS in 7m 15s\n- requirements-check https://zuul.opendev.org/t/openstack/build/e7cd96488e284fa6a516b6b481b49696 : SUCCESS in 3m 20s\n- grenade-aodh https://zuul.opendev.org/t/openstack/build/c7808de46abb46d3ba21bb8951ee96ef : SUCCESS in 1h 08m 28s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/ed1f0c87b76546429dfee98a33226084 : SUCCESS in 55m 36s","accounts_in_message":[],"_revision_number":5},{"id":"c1d5ef2273dbbf80567760e0c2d80a154a572e36","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2025-10-07 15:02:52.000000000","message":"Patch Set 5:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":5},{"id":"c3a097d301dfa4a2b7c9e8ab260bbd266debcd43","author":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"date":"2025-10-15 11:03:37.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"bc8dd1cb9bafe5fac8af09d5eab2ce848f120787","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2025-11-11 08:08:09.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"9716a01d985967a424c280100fb5fcf027bb7efe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-11 09:16:01.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c33e9412496e40a6be5e1670b1cbc2eb\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d777d91dcc9484abc6170450d288045 : SUCCESS in 4m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c0c3a2dc154f4e599792a266c5be50b9 : SUCCESS in 4m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6ea4eb31de6f4faa96e35d2dc359d4bf : SUCCESS in 7m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f002550743b4e6e98d8b941de8e264e : SUCCESS in 8m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6cf7b25fa1e74d97816f60845688f491 : SUCCESS in 7m 10s\n- requirements-check https://zuul.opendev.org/t/openstack/build/c9888f6e7b5f4e0698f3740733a73938 : SUCCESS in 2m 52s\n- grenade-aodh https://zuul.opendev.org/t/openstack/build/a79cc973cbef4022b82f45eeba8262cf : SUCCESS in 1h 02m 53s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/06882b4ffe014be8a556eb6f378eefa3 : SUCCESS in 46m 57s","accounts_in_message":[],"_revision_number":5},{"id":"5a4f37d2608c6730146bfcf1cfa8e6142072fcf7","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2025-11-21 07:12:04.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"b757bbc4cb77db98b1ca68138e8866faa7fd69e3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-21 07:12:30.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"303a49c27d6dd7cf228d9501b5cd44bbeec1c74f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-21 08:25:53.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e59c453978734392b1215601d27a2817\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d04580511ab040dfb23451e9fcd95e03 : SUCCESS in 3m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cc7f6f7ef41f41e2931f50ae2f31ab85 : SUCCESS in 3m 44s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8ebfc3448e5d4398a96a8dae9e45c737 : SUCCESS in 4m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28b72a078ac648ee9c38dd8d4fffdafd : SUCCESS in 8m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9b8d0bea32504488add986242f1de52a : SUCCESS in 7m 15s\n- requirements-check https://zuul.opendev.org/t/openstack/build/5e9e3487262f48e4952411d9367db762 : SUCCESS in 2m 47s\n- grenade-aodh https://zuul.opendev.org/t/openstack/build/196a3714743f4116a8884cac9b100196 : SUCCESS in 1h 08m 50s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/badc80b0d174427c88b3b60c90974bbb : SUCCESS in 30m 01s","accounts_in_message":[],"_revision_number":5},{"id":"ae0f2345bc36c43bca1641fde51168afad1bdb73","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-21 08:25:53.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"d8a448ce4b046b4974841e9bb5d6373416d732c5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-21 08:27:01.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3f4be4df080a41d480c47eff8e8fbf5a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d32ecaa819a3462dbe2397d7cb609c8d : SUCCESS in 39s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1abbca1607754d34a1429b1dd78c4af3 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"23ddf20d7005af4e834f61698cbe32cbc99bb997","revisions":{"5269edc6e136e19654cf80c4918b7d915d3bfd1f":{"kind":"REWORK","_number":1,"created":"2025-08-13 19:23:48.000000000","uploader":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"ref":"refs/changes/79/957279/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/79/957279/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/79/957279/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/79/957279/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/79/957279/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/79/957279/1"}}},"commit":{"parents":[{"commit":"1d7fac6254decc707789ce9121a631f5cee84c6f","subject":"Merge \"Replace independent aodh-api service by Apache web server\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/1d7fac6254decc707789ce9121a631f5cee84c6f"}]}],"author":{"name":"Jaromir Wysoglad","email":"jwysogla@redhat.com","date":"2025-07-30 06:34:05.000000000","tz":-240},"committer":{"name":"Jaromir Wysoglad","email":"jwysogla@redhat.com","date":"2025-08-13 19:04:37.000000000","tz":-240},"subject":"Rbac restrictions to prometheus alarms","message":"Rbac restrictions to prometheus alarms\n\nFor nonadmin users, restrict prometheus alarms to the current project.\nFor admins, restrict the alarms to the specified project (when\nexplicitly specified) or leave them unrestricted. This is very what we already do\nwith Gnocchi alarms in\nhttps://opendev.org/openstack/aodh/src/branch/master/aodh/api/controllers/v2/alarm_rules/gnocchi.py#L181\n\nChange-Id: I6f1b0a1bc67376dca3e00585f04bec697781113c\nSigned-off-by: Jaromir Wysoglad \u003cjwysogla@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/5269edc6e136e19654cf80c4918b7d915d3bfd1f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/5269edc6e136e19654cf80c4918b7d915d3bfd1f"}]},"branch":"refs/heads/master"},"f8eba277ce5530161d7d4a1e0aa85adba1a3cc0b":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2025-08-13 19:24:46.000000000","uploader":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"ref":"refs/changes/79/957279/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/79/957279/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/79/957279/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/79/957279/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/79/957279/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/79/957279/2"}}},"commit":{"parents":[{"commit":"1d7fac6254decc707789ce9121a631f5cee84c6f","subject":"Merge \"Replace independent aodh-api service by Apache web server\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/1d7fac6254decc707789ce9121a631f5cee84c6f"}]}],"author":{"name":"Jaromir Wysoglad","email":"jwysogla@redhat.com","date":"2025-07-30 06:34:05.000000000","tz":-240},"committer":{"name":"Jaromir Wysoglad","email":"jwysogla@redhat.com","date":"2025-08-13 19:06:18.000000000","tz":-240},"subject":"Rbac restrictions to prometheus alarms","message":"Rbac restrictions to prometheus alarms\n\nFor nonadmin users, restrict prometheus alarms to the current project.\nFor admins, restrict the alarms to the specified project (when\nexplicitly specified) or leave them unrestricted. This is very what we already do\nwith Gnocchi alarms in\nhttps://opendev.org/openstack/aodh/src/branch/master/aodh/api/controllers/v2/alarm_rules/gnocchi.py#L181\n\nDepends-On: https://review.opendev.org/c/openstack/requirements/+/956868\nChange-Id: I6f1b0a1bc67376dca3e00585f04bec697781113c\nSigned-off-by: Jaromir Wysoglad \u003cjwysogla@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/f8eba277ce5530161d7d4a1e0aa85adba1a3cc0b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/f8eba277ce5530161d7d4a1e0aa85adba1a3cc0b"}]},"branch":"refs/heads/master"},"e947239f1e70e7ab4d76068c9760a51858b23b34":{"kind":"REWORK","_number":3,"created":"2025-08-15 07:43:40.000000000","uploader":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"ref":"refs/changes/79/957279/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/79/957279/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/79/957279/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/79/957279/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/79/957279/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/79/957279/3"}}},"commit":{"parents":[{"commit":"1d7fac6254decc707789ce9121a631f5cee84c6f","subject":"Merge \"Replace independent aodh-api service by Apache web server\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/1d7fac6254decc707789ce9121a631f5cee84c6f"}]}],"author":{"name":"Jaromir Wysoglad","email":"jwysogla@redhat.com","date":"2025-07-30 06:34:05.000000000","tz":-240},"committer":{"name":"Jaromir Wysoglad","email":"jwysogla@redhat.com","date":"2025-08-15 07:24:59.000000000","tz":-240},"subject":"Rbac restrictions to prometheus alarms","message":"Rbac restrictions to prometheus alarms\n\nFor nonadmin users, restrict prometheus alarms to the current project.\nFor admins, restrict the alarms to the specified project (when\nexplicitly specified) or leave them unrestricted. This is very what we already do\nwith Gnocchi alarms in\nhttps://opendev.org/openstack/aodh/src/branch/master/aodh/api/controllers/v2/alarm_rules/gnocchi.py#L181\n\nDepends-On: https://review.opendev.org/c/openstack/requirements/+/956868\nChange-Id: I6f1b0a1bc67376dca3e00585f04bec697781113c\nSigned-off-by: Jaromir Wysoglad \u003cjwysogla@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/e947239f1e70e7ab4d76068c9760a51858b23b34"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/e947239f1e70e7ab4d76068c9760a51858b23b34"}]},"branch":"refs/heads/master"},"85e1dff0943fea8107a2419f5d8ebfd57e2ee293":{"kind":"REWORK","_number":4,"created":"2025-10-07 12:38:28.000000000","uploader":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"ref":"refs/changes/79/957279/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/79/957279/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/79/957279/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/79/957279/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/79/957279/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/79/957279/4"}}},"commit":{"parents":[{"commit":"1d7fac6254decc707789ce9121a631f5cee84c6f","subject":"Merge \"Replace independent aodh-api service by Apache web server\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/1d7fac6254decc707789ce9121a631f5cee84c6f"}]}],"author":{"name":"Jaromir Wysoglad","email":"jwysogla@redhat.com","date":"2025-07-30 06:34:05.000000000","tz":-240},"committer":{"name":"Jaromir Wysoglad","email":"jwysogla@redhat.com","date":"2025-10-07 12:17:03.000000000","tz":-240},"subject":"Rbac restrictions to prometheus alarms","message":"Rbac restrictions to prometheus alarms\n\nFor nonadmin users, restrict prometheus alarms to the current project.\nFor admins, restrict the alarms to the specified project (when\nexplicitly specified) or leave them unrestricted. This is very similar to\nwhat we already do with Gnocchi alarms in\nhttps://opendev.org/openstack/aodh/src/branch/master/aodh/api/controllers/v2/alarm_rules/gnocchi.py#L181\n\nChange-Id: I6f1b0a1bc67376dca3e00585f04bec697781113c\nSigned-off-by: Jaromir Wysoglad \u003cjwysogla@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/85e1dff0943fea8107a2419f5d8ebfd57e2ee293"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/85e1dff0943fea8107a2419f5d8ebfd57e2ee293"}]},"branch":"refs/heads/master"},"23ddf20d7005af4e834f61698cbe32cbc99bb997":{"kind":"REWORK","_number":5,"created":"2025-10-07 13:21:27.000000000","uploader":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"ref":"refs/changes/79/957279/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/79/957279/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/79/957279/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/79/957279/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/79/957279/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/79/957279/5"}}},"commit":{"parents":[{"commit":"1d7fac6254decc707789ce9121a631f5cee84c6f","subject":"Merge \"Replace independent aodh-api service by Apache web server\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/1d7fac6254decc707789ce9121a631f5cee84c6f"}]}],"author":{"name":"Jaromir Wysoglad","email":"jwysogla@redhat.com","date":"2025-07-30 06:34:05.000000000","tz":-240},"committer":{"name":"Jaromir Wysoglad","email":"jwysogla@redhat.com","date":"2025-10-07 13:00:02.000000000","tz":-240},"subject":"Rbac restrictions to prometheus alarms","message":"Rbac restrictions to prometheus alarms\n\nFor nonadmin users, restrict prometheus alarms to the current project.\nFor admins, restrict the alarms to the specified project (when\nexplicitly specified) or leave them unrestricted. This is very similar to\nwhat we already do with Gnocchi alarms in\nhttps://opendev.org/openstack/aodh/src/branch/master/aodh/api/controllers/v2/alarm_rules/gnocchi.py#L181\n\nChange-Id: I6f1b0a1bc67376dca3e00585f04bec697781113c\nSigned-off-by: Jaromir Wysoglad \u003cjwysogla@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/23ddf20d7005af4e834f61698cbe32cbc99bb997"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/23ddf20d7005af4e834f61698cbe32cbc99bb997"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
