)]}'
{"id":"openstack%2Faodh~958335","triplet_id":"openstack%2Faodh~master~I1a6626b0139b5287a7385348678128acf2644163","project":"openstack/aodh","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-08-26 16:22:50.000000000","reason":"Change was submitted"},"32968":{"account":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"last_update":"2025-08-26 15:28:07.000000000","reason":"\u003cGERRIT_ACCOUNT_32968\u003e replied on the change","reason_account":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"}},"4264":{"account":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"last_update":"2025-08-26 15:01:53.000000000","reason":"removed on reply"},"34975":{"account":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"last_update":"2025-08-26 16:22:50.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I1a6626b0139b5287a7385348678128acf2644163","subject":"Fix GMR feature not actually enabled","status":"MERGED","created":"2025-08-23 15:10:59.000000000","updated":"2025-08-26 16:24:16.000000000","submitted":"2025-08-26 16:22:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"958335","meta_rev_id":"3bc2d478627619a0fe3d8fb7b2a61b506bc5868c","_number":958335,"virtual_id_number":958335,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-08-26 16:22:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"all":[{"value":2,"date":"2025-08-26 15:01:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":2,"date":"2025-08-26 15:28:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"all":[{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":1,"date":"2025-08-26 15:28:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"}],"reviewers":{"REVIEWER":[{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-23 16:02:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-08-26 14:15:33.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"state":"REVIEWER"},{"updated":"2025-08-26 14:15:33.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"state":"REVIEWER"},{"updated":"2025-08-26 14:15:33.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"state":"REVIEWER"}],"messages":[{"id":"daee981aff8226260b53151fb410ce745fdb4f35","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-08-23 15:10:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0d7c0ca28cb2cbb169dc49f6161de2f371f1e968","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-08-23 15:11:33.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"76321c5e76a2d6ceecc2d608f6baebc89dd2fbda","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-08-23 15:13:26.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"f8ac161fb05274df618f7f2502c580e00c405c86","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-08-23 15:14:42.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"650adb37abec8b5b494413404175b78d6739598a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-08-23 15:15:31.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"911b7422e73ae5d71d1379045e39aa5217bb8ccd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-23 16:02:04.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8fc67e8732c94f3d8cdc46abf1a4e234\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6968fe29cb91497f95f7e299915070b6 : SUCCESS in 3m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3f64e6e6b59745e7a673dbbd86d560ac : SUCCESS in 4m 29s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d04278d655d84d9e98bfa3bce8acd3d3 : SUCCESS in 5m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/af5431a18dd6425eb8297b87963ba39c : SUCCESS in 5m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8197bbbdbe7e403fa51693243e956b71 : SUCCESS in 4m 50s\n- grenade-aodh https://zuul.opendev.org/t/openstack/build/f020409971d1475ea15ce2255d51d539 : SUCCESS in 45m 35s (non-voting)\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/00b0eaf5fe5f40758bbf8a25c6367d58 : SUCCESS in 42m 00s","accounts_in_message":[],"_revision_number":5},{"id":"da50b3d5845c50c82e901ae4f5a36cb0b91a2cd0","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2025-08-26 15:01:53.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"6e84964b9d24c9ff22550190b74ad67479f3900f","author":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"date":"2025-08-26 15:28:07.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"02b4600daf6be8a29817dd9d84194de23c4a2edf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-26 15:28:58.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"2d2e1e8afbfdeb89ecc97224716ae586f5284e4c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-26 16:22:50.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f4a7e9bef36142768aba5a6a24b757a5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae59db4d6c814e828b047aa84cdc8c99 : SUCCESS in 2m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/eb701a6e16624838b4ad8f2c19f37a70 : SUCCESS in 3m 08s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/65044ecf185e418d9cb6373c0f7212c7 : SUCCESS in 3m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/651914231b1044d880a85851f54f49d0 : SUCCESS in 4m 30s\n- grenade-aodh https://zuul.opendev.org/t/openstack/build/afd83820ffaa4f8183a2d7c254c3bc4d : SUCCESS in 53m 29s (non-voting)\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/4f0f0397ea794feaa5496e7922d71f7e : SUCCESS in 41m 00s","accounts_in_message":[],"_revision_number":5},{"id":"5c091b257c7210b458811f3316c377aea7151211","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-26 16:22:50.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"3bc2d478627619a0fe3d8fb7b2a61b506bc5868c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-26 16:24:16.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b2ce027d3026451faf4266301b27d11a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/75855661704f409a9138a571edcab9f4 : SUCCESS in 1m 09s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"fe5181923671163def26014ceb10939337b34692","revisions":{"c71687d815dfe2d865b58bd4dd65e1a9567f8188":{"kind":"REWORK","_number":1,"created":"2025-08-23 15:10:59.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/35/958335/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/35/958335/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/35/958335/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/35/958335/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/35/958335/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/35/958335/1"}}},"commit":{"parents":[{"commit":"1d7fac6254decc707789ce9121a631f5cee84c6f","subject":"Merge \"Replace independent aodh-api service by Apache web server\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/1d7fac6254decc707789ce9121a631f5cee84c6f"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-23 15:09:07.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-23 15:10:34.000000000","tz":540},"subject":"Fix GMR feature not actually enabled","message":"Fix GMR feature not actually enabled\n\nThe argv can be None or [], so the if check is never evaluated as True.\n\nChange-Id: I1a6626b0139b5287a7385348678128acf2644163\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/c71687d815dfe2d865b58bd4dd65e1a9567f8188"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/c71687d815dfe2d865b58bd4dd65e1a9567f8188"}]},"branch":"refs/heads/master"},"6f5812f07e785f1019cc3071309333dbec0bd3ab":{"kind":"REWORK","_number":2,"created":"2025-08-23 15:11:33.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/35/958335/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/35/958335/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/35/958335/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/35/958335/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/35/958335/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/35/958335/2"}}},"commit":{"parents":[{"commit":"1d7fac6254decc707789ce9121a631f5cee84c6f","subject":"Merge \"Replace independent aodh-api service by Apache web server\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/1d7fac6254decc707789ce9121a631f5cee84c6f"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-23 15:09:07.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-23 15:11:26.000000000","tz":540},"subject":"Fix GMR feature not actually enabled","message":"Fix GMR feature not actually enabled\n\nThe argv can be None or [], so the if check is never evaluated as True.\n\nChange-Id: I1a6626b0139b5287a7385348678128acf2644163\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/6f5812f07e785f1019cc3071309333dbec0bd3ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/6f5812f07e785f1019cc3071309333dbec0bd3ab"}]},"branch":"refs/heads/master"},"cdb4338c3a6cc222e8bf5f540f6b88b558e45790":{"kind":"REWORK","_number":3,"created":"2025-08-23 15:13:26.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/35/958335/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/35/958335/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/35/958335/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/35/958335/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/35/958335/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/35/958335/3"}}},"commit":{"parents":[{"commit":"1d7fac6254decc707789ce9121a631f5cee84c6f","subject":"Merge \"Replace independent aodh-api service by Apache web server\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/1d7fac6254decc707789ce9121a631f5cee84c6f"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-23 15:09:07.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-23 15:13:15.000000000","tz":540},"subject":"Fix GMR feature not actually enabled","message":"Fix GMR feature not actually enabled\n\nThe argv can be None or [], so the if check is never evaluated as True.\n\nChange-Id: I1a6626b0139b5287a7385348678128acf2644163\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/cdb4338c3a6cc222e8bf5f540f6b88b558e45790"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/cdb4338c3a6cc222e8bf5f540f6b88b558e45790"}]},"branch":"refs/heads/master"},"94e6d90ae731951d41a718f49c2daa26998a22d6":{"kind":"REWORK","_number":4,"created":"2025-08-23 15:14:42.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/35/958335/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/35/958335/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/35/958335/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/35/958335/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/35/958335/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/35/958335/4"}}},"commit":{"parents":[{"commit":"1d7fac6254decc707789ce9121a631f5cee84c6f","subject":"Merge \"Replace independent aodh-api service by Apache web server\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/1d7fac6254decc707789ce9121a631f5cee84c6f"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-23 15:09:07.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-23 15:14:32.000000000","tz":540},"subject":"Fix GMR feature not actually enabled","message":"Fix GMR feature not actually enabled\n\nThe argv can be None or [], so the if check is never evaluated as True.\n\nChange-Id: I1a6626b0139b5287a7385348678128acf2644163\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/94e6d90ae731951d41a718f49c2daa26998a22d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/94e6d90ae731951d41a718f49c2daa26998a22d6"}]},"branch":"refs/heads/master"},"fe5181923671163def26014ceb10939337b34692":{"kind":"REWORK","_number":5,"created":"2025-08-23 15:15:31.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/35/958335/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/35/958335/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/35/958335/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/35/958335/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/35/958335/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/35/958335/5"}}},"commit":{"parents":[{"commit":"1d7fac6254decc707789ce9121a631f5cee84c6f","subject":"Merge \"Replace independent aodh-api service by Apache web server\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/1d7fac6254decc707789ce9121a631f5cee84c6f"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-23 15:09:07.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-23 15:15:23.000000000","tz":540},"subject":"Fix GMR feature not actually enabled","message":"Fix GMR feature not actually enabled\n\nThe argv can be None or [], so the if check is never evaluated as True.\n\nChange-Id: I1a6626b0139b5287a7385348678128acf2644163\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/fe5181923671163def26014ceb10939337b34692"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/fe5181923671163def26014ceb10939337b34692"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
