)]}'
{"id":"openstack%2Faodh~963164","triplet_id":"openstack%2Faodh~master~I6f322df9a72316f97f8e67427d9cbdc4e2c01505","project":"openstack/aodh","branch":"master","topic":"bug/2126581","attention_set":{},"removed_from_attention_set":{"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2025-10-27 17:16:57.000000000","reason":"Change was submitted"},"32968":{"account":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"last_update":"2025-10-27 17:16:57.000000000","reason":"Change was submitted"},"4264":{"account":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"last_update":"2025-10-27 13:39:36.000000000","reason":"\u003cGERRIT_ACCOUNT_4264\u003e replied on the change","reason_account":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"}},"34975":{"account":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"last_update":"2025-10-27 17:16:57.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6f322df9a72316f97f8e67427d9cbdc4e2c01505","subject":"rely on devstack to install uwsgi","status":"MERGED","created":"2025-10-06 10:54:37.000000000","updated":"2025-10-27 17:17:59.000000000","submitted":"2025-10-27 17:16:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"963164-bug/2126581","meta_rev_id":"f5c57d40ef1623378edb2a10980254ad6e587d07","_number":963164,"virtual_id_number":963164,"owner":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-10-27 17:16:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2025-10-23 16:00:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2025-10-27 13:39:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2025-10-27 13:39:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"}],"reviewers":{"REVIEWER":[{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-10-06 11:55:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-10-07 10:36:22.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"},{"updated":"2025-10-23 16:00:03.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2025-10-27 05:17:25.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"state":"REVIEWER"},{"updated":"2025-10-27 05:17:25.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"state":"REVIEWER"},{"updated":"2025-10-27 05:17:25.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"state":"REVIEWER"}],"messages":[{"id":"03702daf3eb6d1035ff8568e319fe9715de0f9e4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-10-06 10:54:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"58f8616b9708d3d513fb6fbf40b0521e8a21a150","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-06 11:55:13.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8ee47b71df5a448593e49a2e66d294a2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1431b177fe404597844bd9a26fb07730 : SUCCESS in 3m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7fa151f978fc46a39c566f03fa24dd05 : SUCCESS in 4m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fd95da0f57f94b8e8fd8bc9f5cad1f79 : SUCCESS in 6m 39s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/641ea3fdfc704ab68bd97abcca4852af : SUCCESS in 5m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/edc207304ce94293a08e2a105de8bef1 : SUCCESS in 9m 37s\n- grenade-aodh https://zuul.opendev.org/t/openstack/build/3b8e1a6b68924bb48b46bab62fdaf205 : FAILURE in 55m 32s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/87356150c439467aa208bc1aac187a29 : SUCCESS in 21m 36s","accounts_in_message":[],"_revision_number":1},{"id":"5561943496d0ae12c697c2a8344edabc99c6030d","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-10-06 17:20:26.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"81758edd0489d6831424630e3c699d0c81a7c70a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-06 18:24:54.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d1eff65622d940a2a134875c6638c27c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/67f257e6f455443c854b52c379d74951 : SUCCESS in 3m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/53a5af8905554917b37d5818d8f9bece : SUCCESS in 4m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a7635a56d30e4997a9031d510adba62f : SUCCESS in 5m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ebcd9c6e1f8f4c198053ba82bcbf7e9f : SUCCESS in 7m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef6147ccb341469ba6afa685e213d082 : SUCCESS in 9m 55s\n- grenade-aodh https://zuul.opendev.org/t/openstack/build/e9d743a7b0ad4ea884816d6c87c02125 : FAILURE in 58m 51s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/f2e83a8b4e464226b0b9d14bdaa41573 : SUCCESS in 45m 19s","accounts_in_message":[],"_revision_number":1},{"id":"cdbc7f27fa8f499da2f41d755de90103ae5377a5","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-10-07 10:36:22.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"eac3e0a10a4ed4b8530088274cbee9210904c120","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-10-23 16:00:03.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"68899cf256c0355004aeefd409cc935effd103d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-23 16:51:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/80734447bb1c4de69f37dda84c52bb3e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58eba5bf1b1a4091b5441dce5c8ea09e : SUCCESS in 2m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/08d0e0139f4945d3bf1917671e2a2549 : SUCCESS in 5m 12s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/071210391aef47988e33c547b39fa539 : SUCCESS in 8m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08237f8ae26149b8913307a094cc54d6 : SUCCESS in 7m 43s\n- grenade-aodh https://zuul.opendev.org/t/openstack/build/6d3bd1e0d30148d2bc55dc71ae5ca564 : SUCCESS in 48m 47s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/e6db99f065c642f9ad6ae748b3be63d0 : SUCCESS in 46m 23s","accounts_in_message":[],"_revision_number":1},{"id":"3cc37f3780ab4a5b13ec600c1f48f5a3cc430caa","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2025-10-27 13:39:36.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d7d5b8be248f6955dc092ba177350408b18395e6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-27 13:40:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"482f97d7950d58df9f66d53edce6ec5a6e80832a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-27 14:40:56.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/97f450028f604ef8978af019aa85c5d1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59d88c40f1aa4004a4603ea934235f13 : SUCCESS in 3m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0b195baf82fd4beb8777dc266c7ccc7e : SUCCESS in 4m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/909e654665c54baa938af75118bb9502 : SUCCESS in 7m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df5141963281492981dd329ad6bbd24f : NODE_FAILURE Node(set) request 681f2a6715ef43488344e4b28e43ac94 failed in 0s\n- grenade-aodh https://zuul.opendev.org/t/openstack/build/c735a3ca135c490295e1d5273344a4d8 : SUCCESS in 56m 20s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/4daaeede23ec427c9012984bad5f4dab : SUCCESS in 49m 53s","accounts_in_message":[],"_revision_number":1},{"id":"bca093bd4d19b2ed3c2687b607b0927c84b76fc7","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-10-27 14:50:32.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c0c2ad98c4ecbd90c701b82045fd5aab09b40e50","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-27 16:01:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/69cd38e3cf2949259d7d55d12bfe5314\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ccb65da5f9d49c782d52a303a30fc68 : SUCCESS in 3m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/73d45f5fa71540b698fa1dd5760e4ad9 : SUCCESS in 2m 55s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/35174acf79b949edac3faa3e9e7673a7 : SUCCESS in 7m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f3ec2c9140444af8a34d78d5ec6181e : SUCCESS in 9m 00s\n- grenade-aodh https://zuul.opendev.org/t/openstack/build/6f9994c1ebe54abfb3f56269e46f2dc3 : SUCCESS in 1h 05m 41s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/163abeb8763248709cd8a25f1766e5ff : SUCCESS in 47m 52s","accounts_in_message":[],"_revision_number":1},{"id":"efea5d39a3620a335af630b26bc2b7eb69595e86","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-27 16:01:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"326b8f3ca70abac4d206bbe6b4b2f2ca3aa44828","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-27 17:16:57.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5806f17ea00243b0887f4fdde4768442\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7084d1d488c747cd80e900fbd505fa2a : SUCCESS in 4m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f895d83bcd054229b4c3806333305b61 : SUCCESS in 5m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1177ae7d222942e493f4893e5baf7403 : SUCCESS in 7m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a643267376154fe2a77f975577aa4971 : SUCCESS in 8m 47s\n- grenade-aodh https://zuul.opendev.org/t/openstack/build/dfb20fbf1f2c40818f639af43c839ed5 : SUCCESS in 1h 10m 10s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/b3e2b822991748c1a12190eaa89a20a0 : SUCCESS in 44m 30s","accounts_in_message":[],"_revision_number":1},{"id":"377c3a2bae9f76a3cbb6bec4cc4e2a186e187e54","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-27 17:16:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"f5c57d40ef1623378edb2a10980254ad6e587d07","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-27 17:17:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b86655d31d26477baf717b4e942d7b86\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e3ee234e0991457a9adb93a6483d0dd8 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"15d38e514d553c906410232ac47b6c6100746b26","revisions":{"15d38e514d553c906410232ac47b6c6100746b26":{"kind":"REWORK","_number":1,"created":"2025-10-06 10:54:37.000000000","uploader":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"ref":"refs/changes/64/963164/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/64/963164/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/64/963164/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/64/963164/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/64/963164/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/64/963164/1"}}},"commit":{"parents":[{"commit":"2f68ba75427b8aa2c5e471f1cf2c6feb993762d3","subject":"Merge \"Remove option for apache \u003c 2.4\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/2f68ba75427b8aa2c5e471f1cf2c6feb993762d3"}]}],"author":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2025-10-06 10:48:58.000000000","tz":60},"committer":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2025-10-06 10:54:09.000000000","tz":60},"subject":"rely on devstack to install uwsgi","message":"rely on devstack to install uwsgi\n\ninstallation of uwsgi from pypi is flaky\nas the compilation of the uwsgi c module sometimes\nfails for non determinisic reasons.\n\nDevstack already will install uwsgi for us and\nit will prefer to use the distro package if possible avoiding\nthe need to compile it in the job. this is both faster and\nmore reliable reducing rechecks.\n\nCloses-Bug: #2126581\nChange-Id: I6f322df9a72316f97f8e67427d9cbdc4e2c01505\nSigned-off-by: Sean Mooney \u003cwork@seanmooney.info\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/15d38e514d553c906410232ac47b6c6100746b26"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/15d38e514d553c906410232ac47b6c6100746b26"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
