)]}'
{"id":"openstack%2Faodh~975395","triplet_id":"openstack%2Faodh~master~I4a3bec69a1cb7a6229a502a048da5f7e428daf8e","project":"openstack/aodh","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-03-11 18:28:13.000000000","reason":"Change was submitted"},"32968":{"account":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"last_update":"2026-03-11 17:06:49.000000000","reason":"\u003cGERRIT_ACCOUNT_32968\u003e replied on the change","reason_account":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"}},"34975":{"account":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"last_update":"2026-02-03 14:18:45.000000000","reason":"\u003cGERRIT_ACCOUNT_34975\u003e replied on the change","reason_account":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"}}},"hashtags":[],"change_id":"I4a3bec69a1cb7a6229a502a048da5f7e428daf8e","subject":"devstack: Manipulate database only where enabled","status":"MERGED","created":"2026-02-02 13:59:35.000000000","updated":"2026-03-11 18:29:19.000000000","submitted":"2026-03-11 18:28:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"975395","meta_rev_id":"85c23c4ad221ce4fbf048a7ac38e1813d611eccc","_number":975395,"virtual_id_number":975395,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-11 18:28:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"all":[{"value":2,"date":"2026-03-11 17:06:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-02-03 14:18:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"all":[{"value":1,"date":"2026-03-11 17:06:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-02 15:21:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-03 10:35:39.000000000","updated_by":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"reviewer":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"state":"REVIEWER"},{"updated":"2026-03-09 15:27:41.000000000","updated_by":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"reviewer":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"state":"REVIEWER"}],"messages":[{"id":"221a468f4626b959d6ca85057f009c0aea123c3c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-02-02 13:59:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e231e16734b2e14a83ff5af1d736a071cd09d17a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-02 15:21:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/412e36f71b2a4b06913027bc1913f147\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e8e2871020504df182e4c822927233e0 : SUCCESS in 3m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/45a6ff5598184c7b9e04fe3f6b28aae1 : SUCCESS in 4m 36s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/28035ded7a1a4b73814a9f8412ce48ca : SUCCESS in 4m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7f2e53c1566407c9f63fe1368523dc2 : SUCCESS in 9m 19s\n- grenade-aodh https://zuul.opendev.org/t/openstack/build/b39c6e8fe60048148c2465716cd7704e : SUCCESS in 1h 12m 41s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/87e2facf7f084331aeafffbcfc2b9388 : SUCCESS in 50m 01s","accounts_in_message":[],"_revision_number":1},{"id":"7b6ba27c3fc0616c244b51dd5ace1648f7f15c2d","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2026-02-03 10:35:39.000000000","message":"Patch Set 1: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":1},{"id":"8360051d69ae222a9e0bec3a0a8929a43ab92791","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-02-03 14:13:50.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"41c0968a28fdfe65e928a99e34bee967341ec317","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-02-03 14:14:20.000000000","message":"Patch Set 1:\n\n(4 comments)","accounts_in_message":[],"_revision_number":1},{"id":"b5e3db9ff7388ed93fa7ccb05a1727a3fbb852d5","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2026-02-03 14:18:45.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"5685b872845c41b2b9735dac3d0140233bf524b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-03 15:18:07.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d48a34010dfc42c39b400435df5bf85d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f62e5c8f01f943e8ae5517ea940f06c6 : SUCCESS in 2m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6a1b1385e2ae41a388bfe6185af56105 : SUCCESS in 5m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/330967e9253a4d27b89455d1eff3e380 : SUCCESS in 4m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2cb14919e6754be3abc137581d669135 : SUCCESS in 7m 42s\n- grenade-aodh https://zuul.opendev.org/t/openstack/build/2d85b8d0cb8d42ce81101f33f8a4d62d : SUCCESS in 59m 35s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/52bc9046f85a40619b5e1db25bf57602 : SUCCESS in 49m 19s","accounts_in_message":[],"_revision_number":2},{"id":"6ee9caa8c023e9709a149b2345433de7e87ec589","author":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"date":"2026-03-11 17:06:49.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"f5f8cc6caa8063d2ab0f830cc63dce5979bcf560","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-11 17:07:20.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"027073e87783db7f90dbf89cbf52fd9838624044","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-11 18:28:13.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/71394e4a861d4dbcb285829e628f1b68\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f3b1fb313d6f4eaf8e6966b57aac77b6 : SUCCESS in 3m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/66f30563a0ad461fb264242a586241ae : SUCCESS in 5m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/043f9d6341fd4e6fbf8cf0eac7d23ad2 : SUCCESS in 4m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81ae99dcfbb3478a812f7c7e7bd52493 : SUCCESS in 9m 37s\n- grenade-aodh https://zuul.opendev.org/t/openstack/build/2c67f45dd7894e929446757f65d34e18 : SUCCESS in 1h 14m 37s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/3dc3838dcb5849dc8cab2e3e26b15ff1 : SUCCESS in 53m 12s","accounts_in_message":[],"_revision_number":2},{"id":"f6ca162b8014cc230ed742a99977dc83725a2634","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-11 18:28:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"85c23c4ad221ce4fbf048a7ac38e1813d611eccc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-11 18:29:19.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/10d7afbbdcb744688d4997df2db24936\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb3e6f4456694f3595457fd689cf7e04 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"5a50e09d596bf500484dae12eb4ae1e7e03e9ac9","revisions":{"6ca6a2291eec0c6975885497ca27e18245eda9c5":{"kind":"REWORK","_number":1,"created":"2026-02-02 13:59:35.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/95/975395/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/95/975395/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/95/975395/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/95/975395/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/95/975395/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/95/975395/1"}}},"commit":{"parents":[{"commit":"78c4ef3b8d66295dd435c10d42bd45c1e7af3b60","subject":"Merge \"pre-commit: Bump versions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/78c4ef3b8d66295dd435c10d42bd45c1e7af3b60"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-02-02 13:58:10.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-02-02 13:58:43.000000000","tz":540},"subject":"devstack: Manipulate databas only where enabled","message":"devstack: Manipulate databas only where enabled\n\n... to avoid duplicated operations in multi node setup.\n\nAlso split the step to create keystone resources to a separate method\nfollowing the structure pattern widely seen in the other plugins.\n\nChange-Id: I4a3bec69a1cb7a6229a502a048da5f7e428daf8e\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/6ca6a2291eec0c6975885497ca27e18245eda9c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/6ca6a2291eec0c6975885497ca27e18245eda9c5"}]},"branch":"refs/heads/master"},"5a50e09d596bf500484dae12eb4ae1e7e03e9ac9":{"kind":"REWORK","_number":2,"created":"2026-02-03 14:13:50.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/95/975395/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/95/975395/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/95/975395/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/95/975395/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/95/975395/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/95/975395/2"}}},"commit":{"parents":[{"commit":"78c4ef3b8d66295dd435c10d42bd45c1e7af3b60","subject":"Merge \"pre-commit: Bump versions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/78c4ef3b8d66295dd435c10d42bd45c1e7af3b60"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-02-02 13:58:10.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-02-03 14:13:39.000000000","tz":540},"subject":"devstack: Manipulate database only where enabled","message":"devstack: Manipulate database only where enabled\n\n... to avoid duplicated operations in multi node setup.\n\nAlso split the step to create keystone resources to a separate method\nfollowing the structure pattern widely seen in the other plugins.\n\nChange-Id: I4a3bec69a1cb7a6229a502a048da5f7e428daf8e\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/5a50e09d596bf500484dae12eb4ae1e7e03e9ac9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/5a50e09d596bf500484dae12eb4ae1e7e03e9ac9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
