)]}'
{"id":"openstack%2Faodh~982765","triplet_id":"openstack%2Faodh~stable%2F2026.1~I6893b3a86ed2418ad214476d1a0494ca82ddf0d8","project":"openstack/aodh","branch":"stable/2026.1","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-03-31 15:09:44.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"I6893b3a86ed2418ad214476d1a0494ca82ddf0d8","subject":"Bump gnocchi version in grenade","status":"MERGED","created":"2026-03-31 10:37:04.000000000","updated":"2026-03-31 17:28:30.000000000","submitted":"2026-03-31 17:26:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"982765","cherry_pick_of_change":982631,"cherry_pick_of_patch_set":1,"meta_rev_id":"522936e4eb340c1828aed4f169bd5e63cd05da06","_number":982765,"virtual_id_number":982765,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-31 17:26:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-03-31 13:48:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-03-31 13:48:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-31 11:17:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-31 13:48:52.000000000","updated_by":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"reviewer":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"state":"REVIEWER"}],"messages":[{"id":"53be8e3b1abcb5010adc93befea12938ce8b62fa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-03-31 10:37:04.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"71ff2edd50ac86d66bb24abffdc86cec534fb862","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-31 11:17:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d3b803a06e00469cbb06f74afed31d31\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b9c26f113c234d8eaf53c0c9126a9d40 : SUCCESS in 4m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/aeccdd9182704385ac6ee423824ee3e6 : SUCCESS in 4m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e9bc6715fd464a10acf128c8d6c0ec27 : SUCCESS in 5m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60c2cd308bfa465ba6560d6212fbde92 : SUCCESS in 8m 34s\n- grenade-aodh https://zuul.opendev.org/t/openstack/build/5b3a4198540741b692462f19046fcce2 : SUCCESS in 36m 48s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/f27f773dc6d345599556265a5d3cf3f7 : SUCCESS in 28m 29s","accounts_in_message":[],"_revision_number":1},{"id":"af71b51d21ac366f2e3894fa0a2f9ab138a49481","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2026-03-31 13:48:52.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"27a958600cba07b2c0a6e4b52912d79a4b5d901f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-31 13:50:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a76f86ed8323c8c32e6e793f493f2aaf453c9f55","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-31 14:36:14.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fe9dae7f9b4c472f9941077241fb8d97\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d072b1718a949a78b48ad0a125d1db3 : SUCCESS in 3m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b3816dc2759e43c1a2edf0d696bb8ca9 : SUCCESS in 4m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/73d7ae45843d4c838219f9854c40bf9a : SUCCESS in 4m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d67ccc5b9184487963912b70f83a9e8 : SUCCESS in 8m 47s\n- grenade-aodh https://zuul.opendev.org/t/openstack/build/b2b78a1599024334ae858554f706d42e : FAILURE in 12m 37s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/273b93a1b07d4bbfb8f6574b3ec012cc : SUCCESS in 45m 10s","accounts_in_message":[],"_revision_number":1},{"id":"727a1f6244d826bde8115d8f01f087e172c1e524","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-03-31 15:09:44.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cbd26e9e91253a4de771da572b4c789845846f53","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-31 16:46:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ed66dc3f43ad421fa29ee92ffd4ad6f6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/09a8745e71dd403ea5ad8961098e41d2 : SUCCESS in 4m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2aca02b752dc415e91376a55aa1baa55 : SUCCESS in 5m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b7d52ba24b58455ebce84c48b0a705cf : SUCCESS in 5m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e11f1aa1e39a4af0b204c819890ddf3b : SUCCESS in 10m 59s\n- grenade-aodh https://zuul.opendev.org/t/openstack/build/51c5e5e1ad38405b8704fffe3973e0e7 : SUCCESS in 1h 27m 40s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/1b7d5fa87b944bdbae7c00715509715d : SUCCESS in 53m 15s","accounts_in_message":[],"_revision_number":1},{"id":"8c368167090481ad3b713c8b26b75405981e9460","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-31 16:46:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d1ed0a53db51682d628fe8b8c22893573100d8ec","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-31 17:26:25.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aae0e0d2c997456f89f894d7bcd63ddb\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7235f50f479f45b1a1d1a6e7763f3ad6 : SUCCESS in 4m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4a41a19fa0fe42c7b96ac96e794a4cf3 : SUCCESS in 4m 34s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e74d46b1136241db9951049f9e4bec83 : SUCCESS in 3m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f71e9b8c5d614e17a312e1191f2da595 : SUCCESS in 9m 38s\n- grenade-aodh https://zuul.opendev.org/t/openstack/build/6d1155dd68a24f0f9f6094395a3ec555 : SUCCESS in 37m 26s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/3169fdf4191d4bd884ff9663b2501cdc : SUCCESS in 38m 49s","accounts_in_message":[],"_revision_number":1},{"id":"52efcdb0f3c62d1f49c6404330ddebbb6a0c2055","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-31 17:26:25.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"522936e4eb340c1828aed4f169bd5e63cd05da06","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-31 17:28:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/993c7b8879f9494cac648ea3d28cb92b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac00448d616f424291c82e9e56b3806e : SUCCESS in 45s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ef38edc511cfa75c6f359ee6a3c04e3d5c71e0d1","revisions":{"ef38edc511cfa75c6f359ee6a3c04e3d5c71e0d1":{"kind":"REWORK","_number":1,"created":"2026-03-31 10:37:04.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/65/982765/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/65/982765/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/65/982765/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/65/982765/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/65/982765/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/65/982765/1"}}},"commit":{"parents":[{"commit":"1c21250732921dfbb892a74fa34a2d41b75d0955","subject":"Update TOX_CONSTRAINTS_FILE for stable/2026.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/1c21250732921dfbb892a74fa34a2d41b75d0955"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-30 11:40:43.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-31 10:37:04.000000000","tz":0},"subject":"Bump gnocchi version in grenade","message":"Bump gnocchi version in grenade\n\nstable/4.7 has been used since 2025.2 .\n\nChange-Id: I6893b3a86ed2418ad214476d1a0494ca82ddf0d8\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n(cherry picked from commit 1200d006d2f8657e678efd43cd6e088c28004153)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/ef38edc511cfa75c6f359ee6a3c04e3d5c71e0d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/ef38edc511cfa75c6f359ee6a3c04e3d5c71e0d1"}]},"branch":"refs/heads/stable/2026.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
