)]}'
{"id":"openstack%2Faodh~983216","triplet_id":"openstack%2Faodh~master~Iba8cf4f650802f5b56558908764187d6ffc58892","project":"openstack/aodh","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-04-21 08:29:39.000000000","reason":"Change was submitted"},"34975":{"account":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"last_update":"2026-04-07 09:17:33.000000000","reason":"\u003cGERRIT_ACCOUNT_34975\u003e replied on the change","reason_account":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"}}},"hashtags":[],"change_id":"Iba8cf4f650802f5b56558908764187d6ffc58892","subject":"Remove app.wsgi","status":"MERGED","created":"2026-04-02 14:49:57.000000000","updated":"2026-04-21 08:31:21.000000000","submitted":"2026-04-21 08:29:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"983216","meta_rev_id":"820620124f07f19b286725b824d9da7f6b0f0ef1","_number":983216,"virtual_id_number":983216,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-21 08:29:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"all":[{"value":2,"date":"2026-04-21 07:12:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-04-03 21:13:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":2,"date":"2026-04-07 09:17:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"all":[{"value":1,"date":"2026-04-21 07:12:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-02 16:09:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-03 21:13:27.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2026-04-03 21:13:27.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"state":"REVIEWER"},{"updated":"2026-04-21 07:12:26.000000000","updated_by":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"reviewer":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"state":"REVIEWER"}],"messages":[{"id":"6b704eddeabfc000cb04f2c42be78b08f82c362b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-04-02 14:49:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"beddec50a1d8b8389b31020b6bfa3de2ab18c148","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-02 16:09:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/08498f43d13c46b78b58782da8c0bf2a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/354a4e051f104558a7e3dae0c8eb73e6 : SUCCESS in 3m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/20957f3f1cbd49e1a9853925deba3117 : SUCCESS in 3m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/99b2faa50fdb4cc68c50ff2647105fb6 : SUCCESS in 2m 59s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/6fa2d939f0b347fe9f18bc82b25ee128 : SUCCESS in 5m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/477937367a834b81b737ac95afaaaa5f : SUCCESS in 8m 24s\n- grenade-aodh https://zuul.opendev.org/t/openstack/build/bca0537f6b104beab2219312a3bd6228 : SUCCESS in 1h 14m 09s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/61c8a9d157af44ca9d42389ad2929c23 : SUCCESS in 41m 39s","accounts_in_message":[],"_revision_number":1},{"id":"35dee02a2fd0e3810b4079cf3872a131de78ced5","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-04-03 21:13:27.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"8afa052dd07aeffc56b092c404af98ca7a5b3c10","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2026-04-07 09:17:33.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"64427cf3b09f4d0398709f25a5330067a8188730","author":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"date":"2026-04-21 07:12:26.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9cae28503b27da3bd7c908f5e8db7867774aa080","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-21 07:13:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b3247f13847ae6fb2bd67b8a1669cabbf318bc8f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-21 08:29:38.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/38687e1e67d74d26b8c8d71fb4ed8582\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/888705efa6054cf7ad5bcde7dc3c307b : SUCCESS in 4m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/606d1fe419d14f6c8812124d71514324 : SUCCESS in 5m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/15f1b5b8320043fd8e7aadac764e4920 : SUCCESS in 3m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c05932cca9146fd841e29bfc121ce2a : SUCCESS in 7m 15s\n- grenade-aodh https://zuul.opendev.org/t/openstack/build/38f72e4e55fe4f219e741d9fc4a8c311 : SUCCESS in 1h 11m 36s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/465bbfc47dbb4faa90ee6f0850bc40a2 : SUCCESS in 51m 09s","accounts_in_message":[],"_revision_number":1},{"id":"7d664371a69d1023d20e868ca35dac00d2d482bb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-21 08:29:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"820620124f07f19b286725b824d9da7f6b0f0ef1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-21 08:31:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b735799e6b1c4103970c4a84f46abb91\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9837ab8c59b7491f873ca6237ca15d3e : SUCCESS in 1m 10s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ae9bf1caa6341b7e03a851015d7854d91f80b1e8","revisions":{"ae9bf1caa6341b7e03a851015d7854d91f80b1e8":{"kind":"REWORK","_number":1,"created":"2026-04-02 14:49:57.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/16/983216/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/16/983216/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/16/983216/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/16/983216/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/16/983216/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/16/983216/1"}}},"commit":{"parents":[{"commit":"bc31232249ff5d523dce12cf9a0554e43e8e7a0d","subject":"Merge \"Remove MAINTAINERS file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/bc31232249ff5d523dce12cf9a0554e43e8e7a0d"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-04-02 14:47:53.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-04-02 14:49:02.000000000","tz":540},"subject":"Remove app.wsgi","message":"Remove app.wsgi\n\nIt was deprecated[1] in favor of the new aodh.wsgi module.\n\n[1] f7aa502e3c62e8fd599e06bdd870a20e8437c90b\n\nChange-Id: Iba8cf4f650802f5b56558908764187d6ffc58892\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/ae9bf1caa6341b7e03a851015d7854d91f80b1e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/ae9bf1caa6341b7e03a851015d7854d91f80b1e8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
