)]}'
{"id":"openstack%2Faodh~992266","triplet_id":"openstack%2Faodh~master~I4175b1c0f961d687a0c015a6b17e834a6f498417","project":"openstack/aodh","branch":"master","attention_set":{},"removed_from_attention_set":{"34975":{"account":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"last_update":"2026-06-09 08:56:47.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4175b1c0f961d687a0c015a6b17e834a6f498417","subject":"Add Aetos jobs","status":"MERGED","created":"2026-06-08 14:29:00.000000000","updated":"2026-06-09 08:58:47.000000000","submitted":"2026-06-09 08:56:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"992266","meta_rev_id":"0329cd166ad28b6d6ba0d233baeaef381c15feb8","_number":992266,"virtual_id_number":992266,"owner":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-09 08:56:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2026-06-08 14:31:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2026-06-08 14:39:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2026-06-08 14:39:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":20676,"name":"daniel.pawlik","display_name":"Daniel Pawlik","email":"dpawlik@redhat.com","username":"daniel.pawlik"}],"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-08 14:31:04.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2026-06-08 14:39:15.000000000","updated_by":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"reviewer":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"state":"REVIEWER"},{"updated":"2026-06-08 15:47:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-09 07:43:11.000000000","updated_by":{"_account_id":20676,"name":"daniel.pawlik","display_name":"Daniel Pawlik","email":"dpawlik@redhat.com","username":"daniel.pawlik"},"reviewer":{"_account_id":20676,"name":"daniel.pawlik","display_name":"Daniel Pawlik","email":"dpawlik@redhat.com","username":"daniel.pawlik"},"state":"CC"}],"messages":[{"id":"acdd2ae1d2b449a8b6c45f8cd69fe5529dad9a2a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2026-06-08 14:29:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7539b21d538ccfb100765f8b70bcad6279b9b4d9","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-06-08 14:31:04.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3190aea3e81590f81afb4e8b545882b6af3981b5","author":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"date":"2026-06-08 14:39:15.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"193f522b96672f75eaf355476aa3c939ce274862","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-08 15:47:35.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/efe21d6f36c7470e9c14614e427363e8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e548d38f42e451a9d8f5f73f9ae8a51 : SUCCESS in 4m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/89aa277e97984389916d819fefd9bb19 : SUCCESS in 4m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/31fc282879e54d0d930b4c1bda20e655 : SUCCESS in 5m 52s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/35f1829bc2fb4038a36e58060a6d0a78 : SUCCESS in 4m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0f3a443143241e090a649ca8b5a6559 : SUCCESS in 7m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cf6de7f95c6449129e8044de173009ba : SUCCESS in 9m 16s\n- grenade-aodh https://zuul.opendev.org/t/openstack/build/00158d6482d349a89ac4bd1945b524e6 : SUCCESS in 1h 15m 46s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/91b4bcdeaf474035b2327c957bc3a252 : SUCCESS in 54m 36s\n- telemetry-dsvm-integration-aetos https://zuul.opendev.org/t/openstack/build/4b975388dcde42438a40646c2b1a8059 : FAILURE in 32m 29s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/d29f3c418e594ffb8d01fa09385cf645 : SUCCESS in 40m 02s\n- telemetry-dsvm-integration-aetos-ipv6-only https://zuul.opendev.org/t/openstack/build/b4ba14be40764fecbf7261b77d55c714 : FAILURE in 57m 20s","accounts_in_message":[],"_revision_number":1},{"id":"08dceb8bb3a2dcda2dcf4df264951c71868b13e1","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2026-06-09 06:42:51.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"350a415efa8d1d1502fe897c14c1349d617b5044","author":{"_account_id":20676,"name":"daniel.pawlik","display_name":"Daniel Pawlik","email":"dpawlik@redhat.com","username":"daniel.pawlik"},"date":"2026-06-09 07:43:11.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3180951e5d7a58c70ce8268d9d33c00eccfddfe9","author":{"_account_id":20676,"name":"daniel.pawlik","display_name":"Daniel Pawlik","email":"dpawlik@redhat.com","username":"daniel.pawlik"},"date":"2026-06-09 07:50:18.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f00f2578d268979fe2bf205a9b6193c281501fd7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-09 07:58:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d82f48a1f0194126b23c427e0b4142ad\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c32c49579bb457cbb78204e8f59ab30 : SUCCESS in 5m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3b8988cc96bb457897c24401fb3ca958 : SUCCESS in 4m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/658cb4494ce24119860ad4dfc345d596 : SUCCESS in 6m 17s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a9c351eda62840f394bc1be76660b050 : SUCCESS in 5m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cfeffea81638462ea2a6e3c0a41ea1c7 : SUCCESS in 8m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fcf6f4ef12b7487e9f437c64b631f75e : SUCCESS in 7m 48s\n- grenade-aodh https://zuul.opendev.org/t/openstack/build/161c3eb405c4488f936697f57eb829d4 : SUCCESS in 1h 09m 14s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/8dd01966accb45819999ca601b1f7a5a : SUCCESS in 49m 44s\n- telemetry-dsvm-integration-aetos https://zuul.opendev.org/t/openstack/build/61902e24f59748b081cb2bae3576d451 : SUCCESS in 38m 44s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/885d01c1da1b4cc2988d6c67a72daa0e : SUCCESS in 51m 05s\n- telemetry-dsvm-integration-aetos-ipv6-only https://zuul.opendev.org/t/openstack/build/4dc25c0fdbc24650867a624b5a6e4613 : SUCCESS in 51m 47s","accounts_in_message":[],"_revision_number":1},{"id":"5233588065a23a89ac1ca279bf5932504a126d74","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-09 07:58:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0c242bb87fd71151bbe7c6e15a53a730d44b2a7c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-09 08:56:47.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/312f9d7d4dbb49e7a495d9532f5a8623\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d6e0141893b4547950133a328f80d70 : SUCCESS in 3m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/549d4156a916493eb1f1cfc16416b2fb : SUCCESS in 4m 34s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1542ebaa87324806aad2dc50f9ae9fc4 : SUCCESS in 4m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c66752fdb4a442fca8ffed6e72ad971b : SUCCESS in 8m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4d3e7841acc64efc846f0c9e723ef0a0 : SUCCESS in 7m 58s\n- grenade-aodh https://zuul.opendev.org/t/openstack/build/9b90c725f7754f5db3285394e9384c29 : SUCCESS in 42m 56s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/2516578b8f0e440f9b85a19d17e57dcd : SUCCESS in 52m 50s\n- telemetry-dsvm-integration-aetos https://zuul.opendev.org/t/openstack/build/c8df18795c6943768301d8868199c58d : SUCCESS in 42m 35s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/a8289ee6a71845d3a6298de61298d0b5 : SUCCESS in 51m 17s\n- telemetry-dsvm-integration-aetos-ipv6-only https://zuul.opendev.org/t/openstack/build/5751779b591b4735ae4b6c9cc9d10ea5 : SUCCESS in 45m 29s","accounts_in_message":[],"_revision_number":1},{"id":"04c6549e443e3f553c1687601c08bc6f8c5fdd6c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-09 08:56:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"0329cd166ad28b6d6ba0d233baeaef381c15feb8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-09 08:58:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1d129fec21bf4c249971830c053d38b4\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1cda3b02795d4cecbef06b8312c68ac8 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c4d9baeeb7909afa47cf4f1ed8548e6b2e12f87e","revisions":{"c4d9baeeb7909afa47cf4f1ed8548e6b2e12f87e":{"kind":"REWORK","_number":1,"created":"2026-06-08 14:29:00.000000000","uploader":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"ref":"refs/changes/66/992266/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/66/992266/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/66/992266/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/66/992266/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/66/992266/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/66/992266/1"}}},"commit":{"parents":[{"commit":"2d54469669fe52de8ab27ecec8dc4d763eba4ced","subject":"Merge \"Align policies with SRBAC goal\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/2d54469669fe52de8ab27ecec8dc4d763eba4ced"}]}],"author":{"name":"Jaromir Wysoglad","email":"jwysogla@redhat.com","date":"2026-06-08 14:28:13.000000000","tz":-240},"committer":{"name":"Jaromir Wysoglad","email":"jwysogla@redhat.com","date":"2026-06-08 14:28:54.000000000","tz":-240},"subject":"Add Aetos jobs","message":"Add Aetos jobs\n\nChange-Id: I4175b1c0f961d687a0c015a6b17e834a6f498417\nSigned-off-by: Jaromir Wysoglad \u003cjwysogla@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/c4d9baeeb7909afa47cf4f1ed8548e6b2e12f87e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/c4d9baeeb7909afa47cf4f1ed8548e6b2e12f87e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
