)]}'
{"id":"openstack%2Faodh~992696","triplet_id":"openstack%2Faodh~master~Icf3fbea1999e968879d98aff3f84910516a0dec2","project":"openstack/aodh","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-06-16 08:27:56.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Icf3fbea1999e968879d98aff3f84910516a0dec2","subject":"Remove Python 3.10 support","status":"MERGED","created":"2026-06-10 15:22:19.000000000","updated":"2026-06-16 08:29:01.000000000","submitted":"2026-06-16 08:27:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"992696","meta_rev_id":"e50e2cb790c62c42e7daa1790cb15cdf940110d9","_number":992696,"virtual_id_number":992696,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-16 08:27:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"all":[{"value":2,"date":"2026-06-10 17:41:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-06-16 07:05:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"all":[{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-06-16 07:05:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-10 15:44:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-10 17:41:49.000000000","updated_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"reviewer":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"state":"REVIEWER"},{"updated":"2026-06-16 07:05:04.000000000","updated_by":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"reviewer":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"state":"REVIEWER"}],"messages":[{"id":"54fdefd02cdfdc5c2566cb37d3a8ae3ecf1c609d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-06-10 15:22:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"22f3aee7f80a28670a52b97b97c4a7155ebaf9f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-10 15:44:32.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7d0c06ca5b7b4eb894a593cf09a5c76b\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/fb22a53f832f4e5da5d3748985e827ec : SUCCESS in 3m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef076c6eacfc4f80b0836d1013148c6e : FAILURE in 5m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/13d2e1cb4158406cbc005c9728652f96 : SUCCESS in 5m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f215bb2d93b14f538e56b61aaeeeebfe : SUCCESS in 5m 05s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/27e9d9680a7b450ab9a924080bd16c0a : SUCCESS in 8m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f37bb00cd7544408c2b6d42d99a8141 : SUCCESS in 9m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2625df5f4f114df1bed63fd8b5bf4ded : SUCCESS in 8m 11s","accounts_in_message":[],"_revision_number":1},{"id":"f9d63c00da62181c43ad24f27a7d01e492b4a5e2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-06-10 15:46:50.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"9a6b2e772a756ec2c2994cb068b7d37eefffaad7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-10 16:52:12.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ef8707a5e26c4e01b228cda9b7c5444b\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/509f12befd5e4397a26b2e3456632c76 : SUCCESS in 3m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d6eaf5e1ea84a9e8822716f6221542c : SUCCESS in 4m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9cad6afa4f394327911ae09ca387d466 : SUCCESS in 5m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/83da571cd1ed4f07aa9c00e82b84459f : SUCCESS in 5m 00s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/4760d73a9d2b4d169710d540644cc65b : SUCCESS in 6m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/005c0be133dc4677aa9b8f63d33910b0 : SUCCESS in 8m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/993ef64a46fb4bd3a065220ffb29c233 : SUCCESS in 7m 27s\n- grenade-aodh https://zuul.opendev.org/t/openstack/build/22d46019a8dc4932a3aed999a935719f : SUCCESS in 1h 00m 27s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/7292b73949b14c83a4cb5e881322daba : SUCCESS in 34m 04s\n- telemetry-dsvm-integration-aetos https://zuul.opendev.org/t/openstack/build/9035b70ad42b48ea90c0ec331bfb6de7 : SUCCESS in 45m 45s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/610d1964539945d392e0f4a69aa875a7 : SUCCESS in 46m 38s\n- telemetry-dsvm-integration-aetos-ipv6-only https://zuul.opendev.org/t/openstack/build/25b8a6b912bd46fbbd3a1ec3092ac34d : SUCCESS in 42m 19s","accounts_in_message":[],"_revision_number":2},{"id":"d931233000b957a0ba0b21b482808c0f12249293","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2026-06-10 17:41:49.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"f152ff09d19db837fea466f89ec66947a53b40d7","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2026-06-16 07:05:04.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"bc9eaf4db5bb95b88b1eb43c386269e88d453460","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-16 07:06:17.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"7cab594e74aaa069053df86b7bdb64633863e868","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-16 08:27:56.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9c7e732db3824ce3bd29e6530d5c3fd8\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/be5c16ba066644c9a842e7453fcca5f7 : SUCCESS in 2m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/44554cd9aa614c408a52876722d690f9 : SUCCESS in 3m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1f0bba3e138d4181a3155a04044f4d0b : SUCCESS in 2m 43s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/05d34d14ca2e4d91970c6687434b6645 : SUCCESS in 3m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a177149e542743ffa7482962d86bff2a : SUCCESS in 8m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/346141e18b864752b88025f47b26c7ca : SUCCESS in 3m 12s\n- grenade-aodh https://zuul.opendev.org/t/openstack/build/338338e732b245efb2217acbd45e17de : SUCCESS in 1h 16m 57s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/bea1d14903364f9f97d606741be0a4bb : SUCCESS in 47m 40s\n- telemetry-dsvm-integration-aetos https://zuul.opendev.org/t/openstack/build/d68f9372dbb94ad3931e1bd309244cc9 : SUCCESS in 26m 07s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/8aad6a726d094ee7bdd8ea141b15d971 : SUCCESS in 36m 18s\n- telemetry-dsvm-integration-aetos-ipv6-only https://zuul.opendev.org/t/openstack/build/b063f227ec9840718d170ddc2af67bc6 : SUCCESS in 22m 58s","accounts_in_message":[],"_revision_number":2},{"id":"024f9fb1dacb056a3739495b005fb65ff1466c1e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-16 08:27:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"e50e2cb790c62c42e7daa1790cb15cdf940110d9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-16 08:29:01.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/872d6c65674b4d60a10d0959aa8c2d76\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95a1408a46a145afa1fde577177bee22 : SUCCESS in 52s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e2fc2e56f0c5e5e756e777a58e645ea61cb76cda","revisions":{"c6194fcf9cb7cd65accdd43397f7be700a86c08d":{"kind":"REWORK","_number":1,"created":"2026-06-10 15:22:19.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/96/992696/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/96/992696/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/96/992696/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/96/992696/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/96/992696/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/96/992696/1"}}},"commit":{"parents":[{"commit":"50ea779dcd6f07d51db4a790edf5e077e9733452","subject":"devstack: Drop outdated comment","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/50ea779dcd6f07d51db4a790edf5e077e9733452"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-06-10 15:21:44.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-06-10 15:22:12.000000000","tz":540},"subject":"Remove Python 3.10 support","message":"Remove Python 3.10 support\n\nPython 3.10 is no longer part of the tested runtimes.\n\nChange-Id: Icf3fbea1999e968879d98aff3f84910516a0dec2\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/c6194fcf9cb7cd65accdd43397f7be700a86c08d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/c6194fcf9cb7cd65accdd43397f7be700a86c08d"}]},"branch":"refs/heads/master"},"e2fc2e56f0c5e5e756e777a58e645ea61cb76cda":{"kind":"REWORK","_number":2,"created":"2026-06-10 15:46:50.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/96/992696/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/96/992696/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/96/992696/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/96/992696/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/96/992696/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/96/992696/2"}}},"commit":{"parents":[{"commit":"50ea779dcd6f07d51db4a790edf5e077e9733452","subject":"devstack: Drop outdated comment","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/50ea779dcd6f07d51db4a790edf5e077e9733452"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-06-10 15:21:44.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-06-10 15:46:39.000000000","tz":540},"subject":"Remove Python 3.10 support","message":"Remove Python 3.10 support\n\nPython 3.10 is no longer part of the tested runtimes.\n\nChange-Id: Icf3fbea1999e968879d98aff3f84910516a0dec2\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/e2fc2e56f0c5e5e756e777a58e645ea61cb76cda"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/e2fc2e56f0c5e5e756e777a58e645ea61cb76cda"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
