)]}'
{"id":"openstack%2Faodh~999503","triplet_id":"openstack%2Faodh~master~I17b17080122d1d09c928ef1c5cbaabec58cb1a18","project":"openstack/aodh","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-08-03 13:26:40.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I17b17080122d1d09c928ef1c5cbaabec58cb1a18","subject":"Remove python3.7 package","status":"MERGED","created":"2026-08-02 09:58:54.000000000","updated":"2026-08-03 13:28:33.000000000","submitted":"2026-08-03 13:26:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"999503","meta_rev_id":"acc2dc6e6a13ff59a594e808835eaeeeb01b6b50","_number":999503,"virtual_id_number":999503,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-08-03 13:26:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"all":[{"value":2,"date":"2026-08-03 06:16:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-08-03 12:07:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"all":[{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-08-03 12:07:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-08-02 11:16:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-08-03 06:16:37.000000000","updated_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"reviewer":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"state":"REVIEWER"},{"updated":"2026-08-03 12:07:41.000000000","updated_by":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"reviewer":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"state":"REVIEWER"}],"messages":[{"id":"da98791da31d7c5229da3d1b4e675733149934de","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-08-02 09:58:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4ba8e09391f60b5dfc6a88f7c0c2b55c3d3165f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-02 11:16:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b18c12c0bf5543689c0887bfa1c3e115\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cebd1415584e472e938302bead3c031c : SUCCESS in 3m 39s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d2dac87e4acf4835b15bb5eeae30519e : SUCCESS in 3m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5b9010cae8f94a22a1d493f6cd135600 : SUCCESS in 4m 18s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/eba2f8b4e5dd4de4946252994b7fb466 : SUCCESS in 3m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62c3030c81e1407f8e9801189e5b7488 : SUCCESS in 9m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0c55995c33264b889e84ece494766bc5 : SUCCESS in 3m 36s\n- grenade-aodh https://zuul.opendev.org/t/openstack/build/7a88e89af8ac44cf83041e62dac02ec0 : SUCCESS in 51m 30s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/a3b9a990227a47dd84015f410961c10b : SUCCESS in 56m 40s\n- telemetry-dsvm-integration-aetos https://zuul.opendev.org/t/openstack/build/044c9616056c40df9709cdfe9d978a41 : SUCCESS in 40m 52s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/ed344b5c2eb54506bb3c1d84eca63713 : SUCCESS in 38m 31s\n- telemetry-dsvm-integration-aetos-ipv6-only https://zuul.opendev.org/t/openstack/build/6ddac8cba2324133bb9e7c2f098d94fa : SUCCESS in 25m 24s","accounts_in_message":[],"_revision_number":1},{"id":"708f878a5d33d9aeb1b901cfc58a6c9b3713f9b2","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2026-08-03 06:16:37.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"bf4940b1ff5b9b0634aaf678c04e3166db747443","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2026-08-03 12:07:41.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2c087b253d4af5fa65a14ea501b5c28cfeb4952d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-03 12:08:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f954d2a7251839f09a34bbb5d9443a3155e2856a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-03 13:26:39.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e51008d4e6564e138cffae1bd8a655b4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a1297c46a784871b173ad3bc4380365 : SUCCESS in 3m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5ef79a5a278c44389bcc0ce040a72e93 : SUCCESS in 3m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/31c123545185497194a582debdfaaa3c : SUCCESS in 2m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b894da6ca9f42d9ba48755e7e0492db : SUCCESS in 4m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/84b3db8b4862484e938f112d6ecc5555 : SUCCESS in 5m 45s\n- grenade-aodh https://zuul.opendev.org/t/openstack/build/15381d22a7c7420d926ee6c47115a3f9 : SUCCESS in 1h 13m 04s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/c5f3f476999e4497aaac8c4b85cd86af : SUCCESS in 30m 07s\n- telemetry-dsvm-integration-aetos https://zuul.opendev.org/t/openstack/build/6cbbc3e5ddaf4079a901f58a17485472 : SUCCESS in 42m 34s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/b911e4c9a84e4af78ddfcbdd13f0ff13 : SUCCESS in 37m 46s\n- telemetry-dsvm-integration-aetos-ipv6-only https://zuul.opendev.org/t/openstack/build/1abc0e7ff83948d2a16d7d4667bd94da : SUCCESS in 48m 15s","accounts_in_message":[],"_revision_number":1},{"id":"7159fa1122293595cfe06f276af94dd77c5b2ee1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-03 13:26:40.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"acc2dc6e6a13ff59a594e808835eaeeeb01b6b50","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-03 13:28:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5162b5cfde58482ea29cc4db66204eb7\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d68325dd6a9b47e39574350ef27b053c : SUCCESS in 1m 42s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/823ba8a0dc9547b8aa70f9899246bb0b : SUCCESS in 47s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b9bb5130ee5957eefe03e6f8a9bf12ba8379e3f3","revisions":{"b9bb5130ee5957eefe03e6f8a9bf12ba8379e3f3":{"kind":"REWORK","_number":1,"created":"2026-08-02 09:58:54.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/03/999503/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/03/999503/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/03/999503/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/03/999503/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/03/999503/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/03/999503/1"}}},"commit":{"parents":[{"commit":"1aeedab0003366db74122bb40eb7331c260b57ae","subject":"Merge \"Clean up deprecated [DEFAULT] heartbeat\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/1aeedab0003366db74122bb40eb7331c260b57ae"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-08-02 09:58:30.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-08-02 09:58:30.000000000","tz":540},"subject":"Remove python3.7 package","message":"Remove python3.7 package\n\nIt\u0027s a leftover from old releases.\n\nChange-Id: I17b17080122d1d09c928ef1c5cbaabec58cb1a18\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/b9bb5130ee5957eefe03e6f8a9bf12ba8379e3f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/b9bb5130ee5957eefe03e6f8a9bf12ba8379e3f3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
