)]}'
{"id":"openstack%2Fapi-site~421688","triplet_id":"openstack%2Fapi-site~master~I74520c7b2673c3a7e742d26c85536f663ce443f1","project":"openstack/api-site","branch":"master","topic":"zanata/translations","hashtags":[],"change_id":"I74520c7b2673c3a7e742d26c85536f663ce443f1","subject":"Imported Translations from Zanata","status":"MERGED","created":"2017-01-18 06:30:20.000000000","updated":"2017-01-18 06:58:36.000000000","submitted":"2017-01-18 06:58:36.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"896a45c83cd1567437290dd111483daeb98bcb63","_number":421688,"virtual_id_number":421688,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":2,"date":"2017-01-18 06:58:36.000000000","post_submit":true,"_account_id":3,"name":"Jenkins","username":"jenkins"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":2,"date":"2017-01-18 06:56:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":1,"date":"2017-01-18 06:56:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2017-01-18 06:56:18.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2017-01-18 06:58:36.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"3dbb0004318cbe4d52325743c6ea168f85f8078e","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2017-01-18 06:30:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a03af09ce5850181895aea7409e5c72546648a1b","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-01-18 06:35:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-api-site-tox-checkniceness-ubuntu-xenial http://logs.openstack.org/88/421688/1/check/gate-api-site-tox-checkniceness-ubuntu-xenial/fbbf121/ : SUCCESS in 2m 19s\n- gate-api-site-tox-doc-publish-checkbuild http://docs-draft.openstack.org/88/421688/1/check/gate-api-site-tox-doc-publish-checkbuild/2761395//publish-docs/ : SUCCESS in 1m 40s\n- gate-api-site-tox-checklang-ubuntu-xenial http://logs.openstack.org/88/421688/1/check/gate-api-site-tox-checklang-ubuntu-xenial/faf43d8/ : SUCCESS in 1m 51s","accounts_in_message":[],"_revision_number":1},{"id":"872f1a0de981097e5970582554a876c164ddbc2b","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2017-01-18 06:56:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4c56b3c9c33af5cf35a9e21879ca6eb915732735","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-01-18 06:56:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":1},{"id":"d596e46a90741f87121735bf03cbb70c259b0460","date":"2017-01-18 06:58:36.000000000","message":"Change has been successfully merged into the git repository by Jenkins","accounts_in_message":[],"_revision_number":1},{"id":"b8944d99df2451b98892a73c5d395b31c51ca6ff","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-01-18 06:58:36.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-api-site-tox-checkniceness-ubuntu-xenial http://logs.openstack.org/88/421688/1/gate/gate-api-site-tox-checkniceness-ubuntu-xenial/8a7fb45/ : SUCCESS in 1m 28s\n- gate-api-site-tox-doc-publish-checkbuild http://docs-draft.openstack.org/88/421688/1/gate/gate-api-site-tox-doc-publish-checkbuild/5803393//publish-docs/ : SUCCESS in 2m 00s\n- gate-api-site-tox-checklang-ubuntu-xenial http://logs.openstack.org/88/421688/1/gate/gate-api-site-tox-checklang-ubuntu-xenial/1300e1d/ : SUCCESS in 1m 59s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d2c43ae7b2092bfbaef70291219bd64fec0f2dcd","revisions":{"d2c43ae7b2092bfbaef70291219bd64fec0f2dcd":{"kind":"REWORK","_number":1,"created":"2017-01-18 06:30:20.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/88/421688/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/api-site","ref":"refs/changes/88/421688/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/api-site refs/changes/88/421688/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/api-site refs/changes/88/421688/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/api-site refs/changes/88/421688/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/api-site refs/changes/88/421688/1"}}},"commit":{"parents":[{"commit":"dfac12405db9c9091181bdceee42ea6a04bd19a8","subject":"Update sphinx version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/api-site/commit/dfac12405db9c9091181bdceee42ea6a04bd19a8"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2017-01-18 06:30:20.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2017-01-18 06:30:20.000000000","tz":0},"subject":"Imported Translations from Zanata","message":"Imported Translations from Zanata\n\nFor more information about this automatic import see:\nhttp://docs.openstack.org/developer/i18n/reviewing-translation-import.html\n\nChange-Id: I74520c7b2673c3a7e742d26c85536f663ce443f1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/api-site/commit/d2c43ae7b2092bfbaef70291219bd64fec0f2dcd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/api-site/commit/d2c43ae7b2092bfbaef70291219bd64fec0f2dcd"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"dfac12405db9c9091181bdceee42ea6a04bd19a8","is_merged_in_target_branch":true,"change_id":"Ibc12cae4692269df2b2bb52d54ee33ce84cb6ed5","change_number":420314,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
