)]}'
{"id":"openstack%2Farch-design~948747","triplet_id":"openstack%2Farch-design~master~I3d703832c8e9e6b69a835343c54a9723d469862f","project":"openstack/arch-design","branch":"master","hashtags":[],"change_id":"I3d703832c8e9e6b69a835343c54a9723d469862f","subject":"wip","status":"ABANDONED","created":"2025-05-03 16:08:46.000000000","updated":"2025-07-20 23:40:20.000000000","total_comment_count":0,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":false,"meta_rev_id":"035b6e06a1e021bd7d467f0f028509a2ae57ad76","_number":948747,"virtual_id_number":948747,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2025-05-03 16:24:36.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"reviewer_updates":[{"updated":"2025-05-03 16:11:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b8c67b0344502c8a41f19ec2ab561692197951fa","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-05-03 16:08:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0596ad271931dbcfba9c2e8528456366671e5e50","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-03 16:11:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f6bf23d91c064bc3b89262ff87d94fec\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21e162c991e448d8aea1a38db0e4f5e8 : SUCCESS in 2m 27s","accounts_in_message":[],"_revision_number":1},{"id":"dc0dd5987878c30fda9109c8f1b22ec272477dbb","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-05-03 16:17:58.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"4a3124c4661b58cb52892e455905d002f2074cfb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-03 16:24:36.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/85fd0deb9aed4bb8b4e815d54f3c8833\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/658c139a3a52403bbae534d5808f2617 : SUCCESS in 2m 21s","accounts_in_message":[],"_revision_number":2},{"id":"035b6e06a1e021bd7d467f0f028509a2ae57ad76","tag":"autogenerated:gerrit:abandon","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-07-20 23:40:20.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"78ce65adb7cc24f0a6eaee615b79bfec03ccccc1","revisions":{"54693a38c7b5f7df88cc4596f37c31531cea1f55":{"kind":"REWORK","_number":1,"created":"2025-05-03 16:08:46.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/47/948747/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/arch-design","ref":"refs/changes/47/948747/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/arch-design refs/changes/47/948747/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/arch-design refs/changes/47/948747/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/arch-design refs/changes/47/948747/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/arch-design refs/changes/47/948747/1"}}},"commit":{"parents":[{"commit":"18e9bab3dedc16e76f9cc84583b4ebe6c5881653","subject":"Merge \"Fix description in OpenStack Logical Architecture figure\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/arch-design/commit/18e9bab3dedc16e76f9cc84583b4ebe6c5881653"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-05-03 16:08:46.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-05-03 16:08:46.000000000","tz":0},"subject":"wip","message":"wip\n\nChange-Id: I3d703832c8e9e6b69a835343c54a9723d469862f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/arch-design/commit/54693a38c7b5f7df88cc4596f37c31531cea1f55"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/arch-design/commit/54693a38c7b5f7df88cc4596f37c31531cea1f55"}]},"branch":"refs/heads/master"},"78ce65adb7cc24f0a6eaee615b79bfec03ccccc1":{"kind":"REWORK","_number":2,"created":"2025-05-03 16:17:58.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/47/948747/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/arch-design","ref":"refs/changes/47/948747/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/arch-design refs/changes/47/948747/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/arch-design refs/changes/47/948747/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/arch-design refs/changes/47/948747/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/arch-design refs/changes/47/948747/2"}}},"commit":{"parents":[{"commit":"18e9bab3dedc16e76f9cc84583b4ebe6c5881653","subject":"Merge \"Fix description in OpenStack Logical Architecture figure\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/arch-design/commit/18e9bab3dedc16e76f9cc84583b4ebe6c5881653"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-05-03 16:08:46.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-05-03 16:17:51.000000000","tz":0},"subject":"wip","message":"wip\n\nChange-Id: I3d703832c8e9e6b69a835343c54a9723d469862f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/arch-design/commit/78ce65adb7cc24f0a6eaee615b79bfec03ccccc1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/arch-design/commit/78ce65adb7cc24f0a6eaee615b79bfec03ccccc1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
