)]}'
{"id":"openstack%2Fastara-appliance~197791","triplet_id":"openstack%2Fastara-appliance~master~I252cf544a468dced1ee7d0d81d5cb55c5bd545d4","project":"openstack/astara-appliance","branch":"master","topic":"bug/1470673","hashtags":[],"change_id":"I252cf544a468dced1ee7d0d81d5cb55c5bd545d4","subject":"follow semver release versioning servers change","status":"MERGED","created":"2015-07-02 00:49:52.000000000","updated":"2015-08-24 18:36:38.000000000","submitted":"2015-08-24 18:36:37.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"8b319cad7cce19863143b0d661b0fd73c7f699ea","_number":197791,"virtual_id_number":197791,"owner":{"_account_id":6923,"name":"Sean Roberts","email":"seanroberts66@gmail.com","username":"seanrob"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":0,"_account_id":2592,"name":"Mark McClain","email":"mark@mcclain.xyz","username":"markmcclain"},{"value":2,"date":"2015-08-24 18:36:37.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":6923,"name":"Sean Roberts","email":"seanroberts66@gmail.com","username":"seanrob"},{"value":0,"_account_id":1420,"name":"Adam Gandelman","email":"adamg@ubuntu.com","username":"gandelman-a"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2592,"name":"Mark McClain","email":"mark@mcclain.xyz","username":"markmcclain"},"all":[{"value":2,"date":"2015-08-24 18:06:00.000000000","_account_id":2592,"name":"Mark McClain","email":"mark@mcclain.xyz","username":"markmcclain"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"date":"2015-08-22 19:41:01.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":6923,"name":"Sean Roberts","email":"seanroberts66@gmail.com","username":"seanrob"},{"value":2,"date":"2015-08-23 20:11:51.000000000","_account_id":1420,"name":"Adam Gandelman","email":"adamg@ubuntu.com","username":"gandelman-a"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2592,"name":"Mark McClain","email":"mark@mcclain.xyz","username":"markmcclain"},"all":[{"value":1,"date":"2015-08-24 18:06:00.000000000","_account_id":2592,"name":"Mark McClain","email":"mark@mcclain.xyz","username":"markmcclain"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":6923,"name":"Sean Roberts","email":"seanroberts66@gmail.com","username":"seanrob"},{"value":0,"_account_id":1420,"name":"Adam Gandelman","email":"adamg@ubuntu.com","username":"gandelman-a"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":1420,"name":"Adam Gandelman","email":"adamg@ubuntu.com","username":"gandelman-a"},{"_account_id":2592,"name":"Mark McClain","email":"mark@mcclain.xyz","username":"markmcclain"},{"_account_id":6923,"name":"Sean Roberts","email":"seanroberts66@gmail.com","username":"seanrob"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2015-08-23 20:11:51.000000000","updated_by":{"_account_id":1420,"name":"Adam Gandelman","email":"adamg@ubuntu.com","username":"gandelman-a"},"reviewer":{"_account_id":1420,"name":"Adam Gandelman","email":"adamg@ubuntu.com","username":"gandelman-a"},"state":"REVIEWER"},{"updated":"2015-08-24 18:06:00.000000000","updated_by":{"_account_id":2592,"name":"Mark McClain","email":"mark@mcclain.xyz","username":"markmcclain"},"reviewer":{"_account_id":2592,"name":"Mark McClain","email":"mark@mcclain.xyz","username":"markmcclain"},"state":"REVIEWER"},{"updated":"2015-08-24 18:36:37.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"65196eba1b9738d9e0c582ee59799e31d0c53ed5","author":{"_account_id":6923,"name":"Sean Roberts","email":"seanroberts66@gmail.com","username":"seanrob"},"date":"2015-07-02 00:49:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f4bf03916114f25752b4cff4538892f7c853953c","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-07-02 01:22:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-akanda-appliance-pep8 http://logs.openstack.org/91/197791/1/check/gate-akanda-appliance-pep8/39fa9dd/ : FAILURE in 13s\n- gate-akanda-appliance-python27 http://logs.openstack.org/91/197791/1/check/gate-akanda-appliance-python27/b1eec85/ : FAILURE in 14s\n- gate-functional-dsvm-akanda-appliance-src http://logs.openstack.org/91/197791/1/check/gate-functional-dsvm-akanda-appliance-src/0749cdd/ : FAILURE in 26m 14s","accounts_in_message":[],"_revision_number":1},{"id":"a30a88632885ba6352beebf79146770e19476969","author":{"_account_id":6923,"name":"Sean Roberts","email":"seanroberts66@gmail.com","username":"seanrob"},"date":"2015-07-26 00:15:09.000000000","message":"Patch Set 1: Code-Review-2\n\nI will go back to late june and tag 2015.2.0. We will review this patch again for LM2.","accounts_in_message":[],"_revision_number":1},{"id":"7935db572d7736c78a27fea6f617676d53d07e64","author":{"_account_id":6923,"name":"Sean Roberts","email":"seanroberts66@gmail.com","username":"seanrob"},"date":"2015-08-22 04:24:05.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9f445f075a162e09110084a6791d985b75d9558a","author":{"_account_id":6923,"name":"Sean Roberts","email":"seanroberts66@gmail.com","username":"seanrob"},"date":"2015-08-22 04:29:56.000000000","message":"Patch Set 2: -Code-Review\n\nmoving to 7.0.0 version","accounts_in_message":[],"_revision_number":2},{"id":"bd71ba23128de0daca506e11f3115166504f6a44","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-08-22 04:49:23.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-akanda-appliance-pep8 http://logs.openstack.org/91/197791/2/check/gate-akanda-appliance-pep8/61449a6/ : FAILURE in 10s\n- gate-akanda-appliance-python27 http://logs.openstack.org/91/197791/2/check/gate-akanda-appliance-python27/d6eac56/ : FAILURE in 12s\n- gate-functional-dsvm-akanda-appliance-src http://logs.openstack.org/91/197791/2/check/gate-functional-dsvm-akanda-appliance-src/eebaaff/ : FAILURE in 25m 11s","accounts_in_message":[],"_revision_number":2},{"id":"6c58548cda5527e9e8a5e0550baea9678941016e","author":{"_account_id":6923,"name":"Sean Roberts","email":"seanroberts66@gmail.com","username":"seanrob"},"date":"2015-08-22 05:18:49.000000000","message":"Patch Set 2: Code-Review-2\n\nforgot to merge a change to move the head off the last tag, hold for that.","accounts_in_message":[],"_revision_number":2},{"id":"339bf491eccc02978f94110d27b53dff08085b10","author":{"_account_id":6923,"name":"Sean Roberts","email":"seanroberts66@gmail.com","username":"seanrob"},"date":"2015-08-22 19:40:27.000000000","message":"Patch Set 2: -Code-Review\n\ntag pushed, ready for this patch","accounts_in_message":[],"_revision_number":2},{"id":"75bdf70d34cf3b67d612e73ab5f9eba3156e4217","author":{"_account_id":6923,"name":"Sean Roberts","email":"seanroberts66@gmail.com","username":"seanrob"},"date":"2015-08-22 19:41:01.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"c63edf086d414e7be9d9b31d025b91afc4a38794","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-08-22 20:12:00.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-akanda-appliance-pep8 http://logs.openstack.org/91/197791/2/check/gate-akanda-appliance-pep8/03cf3e4/ : SUCCESS in 18s\n- gate-akanda-appliance-python27 http://logs.openstack.org/91/197791/2/check/gate-akanda-appliance-python27/c5175ee/ : SUCCESS in 33s\n- gate-functional-dsvm-akanda-appliance-src http://logs.openstack.org/91/197791/2/check/gate-functional-dsvm-akanda-appliance-src/dfb8891/ : SUCCESS in 30m 52s","accounts_in_message":[],"_revision_number":2},{"id":"2bab79d0761a1eaf31ba74613c3bbdafce79c3e6","author":{"_account_id":1420,"name":"Adam Gandelman","email":"adamg@ubuntu.com","username":"gandelman-a"},"date":"2015-08-23 20:11:51.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"2f1b66d6b7fed3fedc47080728058c812822a499","author":{"_account_id":2592,"name":"Mark McClain","email":"mark@mcclain.xyz","username":"markmcclain"},"date":"2015-08-24 18:06:00.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"bd1961f028ce3cbf38ede420e5caffc439235e05","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-08-24 18:06:06.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":2},{"id":"ba6cf41fb3714c312ea7cefe0703025facdec53f","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-08-24 18:36:37.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-akanda-appliance-pep8 http://logs.openstack.org/91/197791/2/gate/gate-akanda-appliance-pep8/bbe186d/ : SUCCESS in 32s\n- gate-akanda-appliance-python27 http://logs.openstack.org/91/197791/2/gate/gate-akanda-appliance-python27/32840d2/ : SUCCESS in 42s\n- gate-functional-dsvm-akanda-appliance-src http://logs.openstack.org/91/197791/2/gate/gate-functional-dsvm-akanda-appliance-src/f23550e/ : SUCCESS in 29m 39s","accounts_in_message":[],"_revision_number":2},{"id":"dd0cd513aa77ee54ebebe890c4cb5e52338a9817","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-08-24 18:36:38.000000000","message":"Change has been successfully merged into the git repository.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c08574ac195880903b7469b5a8e8e61d1c36f225","revisions":{"88cd2857bbcec4ef80b5d5cb6d8783ea89735578":{"kind":"REWORK","_number":1,"created":"2015-07-02 00:49:52.000000000","uploader":{"_account_id":6923,"name":"Sean Roberts","email":"seanroberts66@gmail.com","username":"seanrob"},"ref":"refs/changes/91/197791/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/astara-appliance","ref":"refs/changes/91/197791/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/astara-appliance refs/changes/91/197791/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/astara-appliance refs/changes/91/197791/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/astara-appliance refs/changes/91/197791/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/astara-appliance refs/changes/91/197791/1"}}},"commit":{"parents":[{"commit":"e753bd199e170cd768dad49e80d83d171f611244","subject":"Fix project URL in setup.cfg","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/astara-appliance/commit/e753bd199e170cd768dad49e80d83d171f611244"}]}],"author":{"name":"Sean Roberts","email":"seanroberts66@gmail.com","date":"2015-07-02 00:49:13.000000000","tz":-420},"committer":{"name":"Sean Roberts","email":"seanroberts66@gmail.com","date":"2015-07-02 00:49:13.000000000","tz":-420},"subject":"follow semver release versioning servers change","message":"follow semver release versioning servers change\n\npackaging version will start with the accepted neutron version\nPartial-Bug: #1470673\n\nChange-Id: I252cf544a468dced1ee7d0d81d5cb55c5bd545d4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/astara-appliance/commit/88cd2857bbcec4ef80b5d5cb6d8783ea89735578"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/astara-appliance/commit/88cd2857bbcec4ef80b5d5cb6d8783ea89735578"}]},"branch":"refs/heads/master"},"c08574ac195880903b7469b5a8e8e61d1c36f225":{"kind":"REWORK","_number":2,"created":"2015-08-22 04:24:05.000000000","uploader":{"_account_id":6923,"name":"Sean Roberts","email":"seanroberts66@gmail.com","username":"seanrob"},"ref":"refs/changes/91/197791/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/astara-appliance","ref":"refs/changes/91/197791/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/astara-appliance refs/changes/91/197791/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/astara-appliance refs/changes/91/197791/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/astara-appliance refs/changes/91/197791/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/astara-appliance refs/changes/91/197791/2"}}},"commit":{"parents":[{"commit":"625ec672259aa0b6ad9b326f02ea3a7ed74d99c7","subject":"Don\u0027t truncate interface parsing past the ninth interface.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/astara-appliance/commit/625ec672259aa0b6ad9b326f02ea3a7ed74d99c7"}]}],"author":{"name":"Sean Roberts","email":"seanroberts66@gmail.com","date":"2015-07-02 00:49:13.000000000","tz":-420},"committer":{"name":"Sean Roberts","email":"seanroberts66@gmail.com","date":"2015-08-22 04:24:10.000000000","tz":-420},"subject":"follow semver release versioning servers change","message":"follow semver release versioning servers change\n\npackaging version will start with the accepted neutron version\nPartial-Bug: #1470673\n\nChange-Id: I252cf544a468dced1ee7d0d81d5cb55c5bd545d4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/astara-appliance/commit/c08574ac195880903b7469b5a8e8e61d1c36f225"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/astara-appliance/commit/c08574ac195880903b7469b5a8e8e61d1c36f225"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
