)]}'
{"id":"openstack%2Fastara-neutron~215950","triplet_id":"openstack%2Fastara-neutron~master~I5e41cfcc4f0bab86e9f03548358f58df83589058","project":"openstack/astara-neutron","branch":"master","topic":"bug/1470673","hashtags":[],"change_id":"I5e41cfcc4f0bab86e9f03548358f58df83589058","subject":"follow semver release versioning servers change","status":"MERGED","created":"2015-08-22 18:16:19.000000000","updated":"2015-08-24 18:51:25.000000000","submitted":"2015-08-24 18:51:24.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d139ee8a6225968db0dc9f7d329b23600fcbe723","_number":215950,"virtual_id_number":215950,"owner":{"_account_id":6923,"name":"Sean Roberts","email":"seanroberts66@gmail.com","username":"seanrob"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":0,"_account_id":2592,"name":"Mark McClain","email":"mark@mcclain.xyz","username":"markmcclain"},{"value":2,"date":"2015-08-24 18:51:24.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":1420,"name":"Adam Gandelman","email":"adamg@ubuntu.com","username":"gandelman-a"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2592,"name":"Mark McClain","email":"mark@mcclain.xyz","username":"markmcclain"},"all":[{"value":2,"date":"2015-08-24 18:05:50.000000000","_account_id":2592,"name":"Mark McClain","email":"mark@mcclain.xyz","username":"markmcclain"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":2,"date":"2015-08-23 20:11:37.000000000","_account_id":1420,"name":"Adam Gandelman","email":"adamg@ubuntu.com","username":"gandelman-a"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2592,"name":"Mark McClain","email":"mark@mcclain.xyz","username":"markmcclain"},"all":[{"value":1,"date":"2015-08-24 18:05:50.000000000","_account_id":2592,"name":"Mark McClain","email":"mark@mcclain.xyz","username":"markmcclain"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":1420,"name":"Adam Gandelman","email":"adamg@ubuntu.com","username":"gandelman-a"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":1420,"name":"Adam Gandelman","email":"adamg@ubuntu.com","username":"gandelman-a"},{"_account_id":2592,"name":"Mark McClain","email":"mark@mcclain.xyz","username":"markmcclain"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2015-08-23 20:11:37.000000000","updated_by":{"_account_id":1420,"name":"Adam Gandelman","email":"adamg@ubuntu.com","username":"gandelman-a"},"reviewer":{"_account_id":1420,"name":"Adam Gandelman","email":"adamg@ubuntu.com","username":"gandelman-a"},"state":"REVIEWER"},{"updated":"2015-08-24 18:05:50.000000000","updated_by":{"_account_id":2592,"name":"Mark McClain","email":"mark@mcclain.xyz","username":"markmcclain"},"reviewer":{"_account_id":2592,"name":"Mark McClain","email":"mark@mcclain.xyz","username":"markmcclain"},"state":"REVIEWER"},{"updated":"2015-08-24 18:51:24.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"6266e55142520e48b50961b686dba0ca33f4820a","author":{"_account_id":6923,"name":"Sean Roberts","email":"seanroberts66@gmail.com","username":"seanrob"},"date":"2015-08-22 18:16:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"04cf4f232ed1b0ee078b2a1932d3647a562ec680","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-08-22 18:46:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-akanda-neutron-pep8 http://logs.openstack.org/50/215950/1/check/gate-akanda-neutron-pep8/7c45b43/ : SUCCESS in 33s\n- gate-akanda-neutron-python27 http://logs.openstack.org/50/215950/1/check/gate-akanda-neutron-python27/277464f/ : SUCCESS in 31s\n- gate-functional-dsvm-akanda http://logs.openstack.org/50/215950/1/check/gate-functional-dsvm-akanda/aa30b0e/ : SUCCESS in 29m 38s","accounts_in_message":[],"_revision_number":1},{"id":"628882ad52543fdd3382bf512b174b773764c88e","author":{"_account_id":1420,"name":"Adam Gandelman","email":"adamg@ubuntu.com","username":"gandelman-a"},"date":"2015-08-23 20:11:37.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"555ea41b3f7a318c4681819fab923d0362413c42","author":{"_account_id":2592,"name":"Mark McClain","email":"mark@mcclain.xyz","username":"markmcclain"},"date":"2015-08-24 18:05:50.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f2004c334cc19c58d162ff35b2293cb733e367ec","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-08-24 18:06:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":1},{"id":"8417067e3b15bff249f9b43a0a3cde1b0b990fc4","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-08-24 18:51:24.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-akanda-neutron-pep8 http://logs.openstack.org/50/215950/1/gate/gate-akanda-neutron-pep8/38b6f37/ : SUCCESS in 35s\n- gate-akanda-neutron-python27 http://logs.openstack.org/50/215950/1/gate/gate-akanda-neutron-python27/7a34a2d/ : SUCCESS in 43s\n- gate-functional-dsvm-akanda http://logs.openstack.org/50/215950/1/gate/gate-functional-dsvm-akanda/6026dbc/ : SUCCESS in 38m 04s","accounts_in_message":[],"_revision_number":1},{"id":"1636dde1ff0132aff2b8e5210807e8363aba5941","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-08-24 18:51:25.000000000","message":"Change has been successfully merged into the git repository.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a9abe316c2ef4fe433037fedcbd7ab27b32e9196","revisions":{"a9abe316c2ef4fe433037fedcbd7ab27b32e9196":{"kind":"REWORK","_number":1,"created":"2015-08-22 18:16:19.000000000","uploader":{"_account_id":6923,"name":"Sean Roberts","email":"seanroberts66@gmail.com","username":"seanrob"},"ref":"refs/changes/50/215950/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/astara-neutron","ref":"refs/changes/50/215950/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/astara-neutron refs/changes/50/215950/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/astara-neutron refs/changes/50/215950/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/astara-neutron refs/changes/50/215950/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/astara-neutron refs/changes/50/215950/1"}}},"commit":{"parents":[{"commit":"7b03eb97d3473f8277e08ddadded71ac18b5fe96","subject":"Drop unused Neutron extensions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/astara-neutron/commit/7b03eb97d3473f8277e08ddadded71ac18b5fe96"}]}],"author":{"name":"Sean Roberts","email":"seanroberts66@gmail.com","date":"2015-08-22 18:15:55.000000000","tz":-420},"committer":{"name":"Sean Roberts","email":"seanroberts66@gmail.com","date":"2015-08-22 18:15:55.000000000","tz":-420},"subject":"follow semver release versioning servers change","message":"follow semver release versioning servers change\n\npackaging version will start with the accepted neutron version\nPartial-Bug: #1470673\n\nChange-Id: I5e41cfcc4f0bab86e9f03548358f58df83589058\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/astara-neutron/commit/a9abe316c2ef4fe433037fedcbd7ab27b32e9196"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/astara-neutron/commit/a9abe316c2ef4fe433037fedcbd7ab27b32e9196"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
