)]}'
{"id":"openstack%2Fautomaton~665997","triplet_id":"openstack%2Fautomaton~master~Ied4095b176b20795f6a8d4d7a84d01d958d76896","project":"openstack/automaton","branch":"master","hashtags":[],"change_id":"Ied4095b176b20795f6a8d4d7a84d01d958d76896","subject":"Sync Sphinx requirement","status":"MERGED","created":"2019-06-18 14:29:29.000000000","updated":"2019-07-03 14:55:23.000000000","submitted":"2019-07-03 14:55:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"665997-1562165723334-6eef3a3a","meta_rev_id":"cd93cd10efcf60c48417abfbf38dddb779c7760d","_number":665997,"virtual_id_number":665997,"owner":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"value":0,"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"date":"2019-07-03 11:19:14.000000000","_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"date":"2019-06-28 15:48:40.000000000","_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":2,"date":"2019-07-03 14:55:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},"recommended":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"all":[{"value":1,"date":"2019-06-21 06:44:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"value":2,"date":"2019-06-26 08:09:41.000000000","_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"value":1,"date":"2019-06-21 02:52:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":2,"date":"2019-06-19 16:06:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},"all":[{"value":0,"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"value":1,"date":"2019-06-26 08:09:41.000000000","_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"value":0,"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-19 16:06:34.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2019-06-21 02:52:27.000000000","updated_by":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"reviewer":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"state":"REVIEWER"},{"updated":"2019-06-21 06:44:04.000000000","updated_by":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"reviewer":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"state":"REVIEWER"},{"updated":"2019-06-26 08:09:41.000000000","updated_by":{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},"reviewer":{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},"state":"REVIEWER"},{"updated":"2019-07-03 11:19:14.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2019-07-03 14:55:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e8a8b81d2985cca38e2602a3b89eb5f7fa899c55","author":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"date":"2019-06-18 14:29:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"14f9fd5bcd2580ed3245d0a197bdb625473ad825","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-06-18 15:00:25.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"36a2ba18e3a2e45bca847d481a16028f2c195e64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 16:42:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/97/665997/1/check/requirements-check/a4f2661/ : FAILURE in 4m 27s\n- tempest-full http://logs.openstack.org/97/665997/1/check/tempest-full/f0d64cd/ : SUCCESS in 1h 49m 53s\n- tempest-full-py3 http://logs.openstack.org/97/665997/1/check/tempest-full-py3/b90da4e/ : SUCCESS in 1h 30m 41s\n- openstack-tox-lower-constraints http://logs.openstack.org/97/665997/1/check/openstack-tox-lower-constraints/1c43110/ : SUCCESS in 5m 41s\n- openstack-tox-pep8 http://logs.openstack.org/97/665997/1/check/openstack-tox-pep8/9b9203b/ : SUCCESS in 4m 47s\n- openstack-tox-py27 http://logs.openstack.org/97/665997/1/check/openstack-tox-py27/5fc4c3d/ : SUCCESS in 4m 47s\n- openstack-tox-py35 http://logs.openstack.org/97/665997/1/check/openstack-tox-py35/87b5df5/ : SUCCESS in 4m 45s\n- openstack-tox-py36 http://logs.openstack.org/97/665997/1/check/openstack-tox-py36/c25fec2/ : SUCCESS in 4m 53s\n- openstack-tox-py37 http://logs.openstack.org/97/665997/1/check/openstack-tox-py37/8486911/ : SUCCESS in 5m 16s\n- openstack-tox-docs http://logs.openstack.org/97/665997/1/check/openstack-tox-docs/2b5be3a/html/ : SUCCESS in 4m 41s\n- build-openstack-releasenotes http://logs.openstack.org/97/665997/1/check/build-openstack-releasenotes/6785e54/html/ : SUCCESS in 4m 37s","accounts_in_message":[],"_revision_number":1},{"id":"aadfe151681b92e127e004b36db7525aa014221b","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2019-06-18 16:57:57.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"e8e301ae1267239def33e62e8080b3fe2b4433c0","author":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"date":"2019-06-18 16:57:58.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"cc043512fffded27dbe2868cec96acb01e87cf76","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 18:46:31.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check http://logs.openstack.org/97/665997/2/check/requirements-check/3d09edb/ : SUCCESS in 3m 32s\n- tempest-full http://logs.openstack.org/97/665997/2/check/tempest-full/74cbfd0/ : SUCCESS in 1h 41m 59s\n- tempest-full-py3 http://logs.openstack.org/97/665997/2/check/tempest-full-py3/1877bea/ : SUCCESS in 1h 41m 58s\n- openstack-tox-lower-constraints http://logs.openstack.org/97/665997/2/check/openstack-tox-lower-constraints/30be462/ : SUCCESS in 5m 17s\n- openstack-tox-pep8 http://logs.openstack.org/97/665997/2/check/openstack-tox-pep8/853cdf6/ : SUCCESS in 4m 28s\n- openstack-tox-py27 http://logs.openstack.org/97/665997/2/check/openstack-tox-py27/0aa69b5/ : SUCCESS in 4m 38s\n- openstack-tox-py35 http://logs.openstack.org/97/665997/2/check/openstack-tox-py35/00e7c49/ : SUCCESS in 4m 28s\n- openstack-tox-py36 http://logs.openstack.org/97/665997/2/check/openstack-tox-py36/cb01dd8/ : SUCCESS in 4m 06s\n- openstack-tox-py37 http://logs.openstack.org/97/665997/2/check/openstack-tox-py37/67ccc77/ : SUCCESS in 6m 03s\n- openstack-tox-docs http://logs.openstack.org/97/665997/2/check/openstack-tox-docs/af3deff/html/ : SUCCESS in 4m 15s\n- build-openstack-releasenotes http://logs.openstack.org/97/665997/2/check/build-openstack-releasenotes/a5cbc8f/html/ : SUCCESS in 4m 24s","accounts_in_message":[],"_revision_number":2},{"id":"9cefc0a4508d6c72bf0e381ebef0ebb73944c646","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-06-19 16:06:34.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"87a36b6488ecb1ae338789514e5d868b13be2715","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2019-06-21 02:52:27.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"dd2f0f2458ffa76d9f52b282d43dc394506c411f","author":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"date":"2019-06-21 06:44:04.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"7529490c57a2e329dcfb2387b7d42be6b2516e73","author":{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},"date":"2019-06-26 08:09:41.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"d79eb9092e566dede54c17491c86f618f4ccc270","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-26 08:09:53.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"0c74c2bfa91e204c1b6aacc19975b6a7cc90d433","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-26 10:16:22.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/97/665997/2/gate/requirements-check/7439662/ : SUCCESS in 2m 20s\n- tempest-full http://logs.openstack.org/97/665997/2/gate/tempest-full/d2e0fc0/ : SUCCESS in 1h 37m 38s\n- tempest-full-py3 http://logs.openstack.org/97/665997/2/gate/tempest-full-py3/4cc687a/ : POST_FAILURE in 2h 01m 18s\n- openstack-tox-lower-constraints http://logs.openstack.org/97/665997/2/gate/openstack-tox-lower-constraints/7b91377/ : SUCCESS in 3m 48s\n- openstack-tox-pep8 http://logs.openstack.org/97/665997/2/gate/openstack-tox-pep8/9fe9d26/ : SUCCESS in 2m 52s\n- openstack-tox-py27 http://logs.openstack.org/97/665997/2/gate/openstack-tox-py27/7d1843b/ : SUCCESS in 3m 00s\n- openstack-tox-py35 http://logs.openstack.org/97/665997/2/gate/openstack-tox-py35/11a66f2/ : SUCCESS in 2m 52s\n- openstack-tox-py36 http://logs.openstack.org/97/665997/2/gate/openstack-tox-py36/1a420c0/ : SUCCESS in 2m 49s\n- openstack-tox-py37 http://logs.openstack.org/97/665997/2/gate/openstack-tox-py37/94cf75f/ : SUCCESS in 3m 20s\n- openstack-tox-docs http://logs.openstack.org/97/665997/2/gate/openstack-tox-docs/639c07a/ : FAILURE in 2m 52s\n- build-openstack-releasenotes http://logs.openstack.org/97/665997/2/gate/build-openstack-releasenotes/7c99f85/html/ : SUCCESS in 3m 07s","accounts_in_message":[],"_revision_number":2},{"id":"3f9d9ac5534cbbaf9a8d9c5bbc9f6ce74ab14818","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-06-26 10:51:37.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"0acac0883cbb9b24ba3c67cd847aa3ef53ebe0d4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-26 13:04:28.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/97/665997/2/check/requirements-check/9ded4f9/ : SUCCESS in 2m 20s\n- tempest-full http://logs.openstack.org/97/665997/2/check/tempest-full/2f8f817/ : TIMED_OUT in 2h 06m 21s\n- tempest-full-py3 http://logs.openstack.org/97/665997/2/check/tempest-full-py3/9874263/ : SUCCESS in 1h 36m 45s\n- openstack-tox-lower-constraints http://logs.openstack.org/97/665997/2/check/openstack-tox-lower-constraints/e3580f0/ : FAILURE in 3m 25s\n- openstack-tox-pep8 http://logs.openstack.org/97/665997/2/check/openstack-tox-pep8/4d6521d/ : SUCCESS in 2m 35s\n- openstack-tox-py27 http://logs.openstack.org/97/665997/2/check/openstack-tox-py27/4a84a8b/ : FAILURE in 3m 05s\n- openstack-tox-py35 http://logs.openstack.org/97/665997/2/check/openstack-tox-py35/203ab6e/ : SUCCESS in 3m 17s\n- openstack-tox-py36 http://logs.openstack.org/97/665997/2/check/openstack-tox-py36/0c272db/ : SUCCESS in 3m 25s\n- openstack-tox-py37 http://logs.openstack.org/97/665997/2/check/openstack-tox-py37/aa88004/ : SUCCESS in 5m 04s\n- openstack-tox-docs http://logs.openstack.org/97/665997/2/check/openstack-tox-docs/888aecf/ : FAILURE in 2m 33s\n- build-openstack-releasenotes http://logs.openstack.org/97/665997/2/check/build-openstack-releasenotes/df6696c/html/ : SUCCESS in 3m 15s","accounts_in_message":[],"_revision_number":2},{"id":"b019b3cad4e2873333543210a3f0f7fb659dcb2b","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2019-06-27 00:53:45.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"f2076b3e389d16aa546417d86f2b8c3ef15ebd54","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 02:59:57.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/97/665997/2/check/requirements-check/14d0868/ : SUCCESS in 2m 05s\n- tempest-full http://logs.openstack.org/97/665997/2/check/tempest-full/0c147c2/ : SUCCESS in 2h 03m 06s\n- tempest-full-py3 http://logs.openstack.org/97/665997/2/check/tempest-full-py3/b12479c/ : SUCCESS in 1h 30m 22s\n- openstack-tox-lower-constraints http://logs.openstack.org/97/665997/2/check/openstack-tox-lower-constraints/0d534c3/ : SUCCESS in 4m 10s\n- openstack-tox-pep8 http://logs.openstack.org/97/665997/2/check/openstack-tox-pep8/f0d6e66/ : SUCCESS in 2m 56s\n- openstack-tox-py27 http://logs.openstack.org/97/665997/2/check/openstack-tox-py27/33aed55/ : SUCCESS in 2m 59s\n- openstack-tox-py35 http://logs.openstack.org/97/665997/2/check/openstack-tox-py35/b284780/ : SUCCESS in 2m 52s\n- openstack-tox-py36 http://logs.openstack.org/97/665997/2/check/openstack-tox-py36/4e0e769/ : SUCCESS in 2m 48s\n- openstack-tox-py37 http://logs.openstack.org/97/665997/2/check/openstack-tox-py37/0de1d5f/ : SUCCESS in 3m 28s\n- openstack-tox-docs http://logs.openstack.org/97/665997/2/check/openstack-tox-docs/a215c8b/ : FAILURE in 2m 41s\n- build-openstack-releasenotes http://logs.openstack.org/97/665997/2/check/build-openstack-releasenotes/268bd00/html/ : SUCCESS in 3m 16s","accounts_in_message":[],"_revision_number":2},{"id":"cebb6441ac98e7b9eb18935078acf9791b3b1266","author":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"date":"2019-06-28 15:48:40.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"05155a4a683fd92ccb1eb07c05d08a3ff9112243","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-28 17:39:46.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/97/665997/2/check/requirements-check/8e8c39b/ : SUCCESS in 2m 17s\n- tempest-full http://logs.openstack.org/97/665997/2/check/tempest-full/bb3e948/ : SUCCESS in 1h 42m 11s\n- tempest-full-py3 http://logs.openstack.org/97/665997/2/check/tempest-full-py3/c82b1db/ : SUCCESS in 1h 45m 37s\n- openstack-tox-lower-constraints http://logs.openstack.org/97/665997/2/check/openstack-tox-lower-constraints/edde820/ : SUCCESS in 3m 52s\n- openstack-tox-pep8 http://logs.openstack.org/97/665997/2/check/openstack-tox-pep8/2e0dbf5/ : SUCCESS in 2m 48s\n- openstack-tox-py27 http://logs.openstack.org/97/665997/2/check/openstack-tox-py27/a486f5d/ : SUCCESS in 2m 52s\n- openstack-tox-py35 http://logs.openstack.org/97/665997/2/check/openstack-tox-py35/e9e18ab/ : SUCCESS in 3m 12s\n- openstack-tox-py36 http://logs.openstack.org/97/665997/2/check/openstack-tox-py36/6fcb1a4/ : SUCCESS in 2m 53s\n- openstack-tox-py37 http://logs.openstack.org/97/665997/2/check/openstack-tox-py37/3df51da/ : SUCCESS in 3m 05s\n- openstack-tox-docs http://logs.openstack.org/97/665997/2/check/openstack-tox-docs/c4993a1/ : FAILURE in 3m 10s\n- build-openstack-releasenotes http://logs.openstack.org/97/665997/2/check/build-openstack-releasenotes/e6a5fb6/html/ : SUCCESS in 3m 54s","accounts_in_message":[],"_revision_number":2},{"id":"0b3ec2552736cf8c027b7b83796f6d2e1ae07f70","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2019-07-03 11:19:14.000000000","message":"Patch Set 2:\n\nrecheck\n\nbindep patch [1] merged on master, so this patch should be ok now.\n\n[1] Ib9a720379ca997f848212eeca68ed9f32e4ed349","accounts_in_message":[],"_revision_number":2},{"id":"ba5a65addd4307af15c39a81f80ee9fa4c14ed4d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-03 13:07:36.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check http://logs.openstack.org/97/665997/2/check/requirements-check/4be2ea3/ : SUCCESS in 1m 57s\n- tempest-full http://logs.openstack.org/97/665997/2/check/tempest-full/7a8a96b/ : SUCCESS in 1h 37m 38s\n- tempest-full-py3 http://logs.openstack.org/97/665997/2/check/tempest-full-py3/b84b2e8/ : SUCCESS in 1h 42m 37s\n- openstack-tox-lower-constraints http://logs.openstack.org/97/665997/2/check/openstack-tox-lower-constraints/849d376/ : SUCCESS in 4m 17s\n- openstack-tox-pep8 http://logs.openstack.org/97/665997/2/check/openstack-tox-pep8/de2a45d/ : SUCCESS in 3m 44s\n- openstack-tox-py27 http://logs.openstack.org/97/665997/2/check/openstack-tox-py27/a344837/ : SUCCESS in 4m 18s\n- openstack-tox-py35 http://logs.openstack.org/97/665997/2/check/openstack-tox-py35/7b7a8fa/ : SUCCESS in 3m 48s\n- openstack-tox-py36 http://logs.openstack.org/97/665997/2/check/openstack-tox-py36/b4dca47/ : SUCCESS in 3m 47s\n- openstack-tox-py37 http://logs.openstack.org/97/665997/2/check/openstack-tox-py37/e59a20d/ : SUCCESS in 3m 27s\n- openstack-tox-docs http://logs.openstack.org/97/665997/2/check/openstack-tox-docs/2ecf030/html/ : SUCCESS in 2m 53s\n- build-openstack-releasenotes http://logs.openstack.org/97/665997/2/check/build-openstack-releasenotes/9995bcf/html/ : SUCCESS in 4m 10s","accounts_in_message":[],"_revision_number":2},{"id":"a37453139906b6a6ca6f6954c57895363da80b7b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-03 13:07:48.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f3a52b7302508c3a724899d7d01a4311c6e939ae","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-03 14:55:23.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check http://logs.openstack.org/97/665997/2/gate/requirements-check/e16ec8d/ : SUCCESS in 2m 38s\n- tempest-full http://logs.openstack.org/97/665997/2/gate/tempest-full/163cade/ : SUCCESS in 1h 42m 41s\n- tempest-full-py3 http://logs.openstack.org/97/665997/2/gate/tempest-full-py3/7c6cf6c/ : SUCCESS in 1h 31m 22s\n- openstack-tox-lower-constraints http://logs.openstack.org/97/665997/2/gate/openstack-tox-lower-constraints/a9d97d8/ : SUCCESS in 4m 25s\n- openstack-tox-pep8 http://logs.openstack.org/97/665997/2/gate/openstack-tox-pep8/6810dcb/ : SUCCESS in 4m 32s\n- openstack-tox-py27 http://logs.openstack.org/97/665997/2/gate/openstack-tox-py27/b20b700/ : SUCCESS in 3m 00s\n- openstack-tox-py35 http://logs.openstack.org/97/665997/2/gate/openstack-tox-py35/3c3db02/ : SUCCESS in 3m 13s\n- openstack-tox-py36 http://logs.openstack.org/97/665997/2/gate/openstack-tox-py36/3976106/ : SUCCESS in 5m 03s\n- openstack-tox-py37 http://logs.openstack.org/97/665997/2/gate/openstack-tox-py37/452a2cd/ : SUCCESS in 4m 35s\n- openstack-tox-docs http://logs.openstack.org/97/665997/2/gate/openstack-tox-docs/88e25e1/html/ : SUCCESS in 3m 44s\n- build-openstack-releasenotes http://logs.openstack.org/97/665997/2/gate/build-openstack-releasenotes/5de98db/html/ : SUCCESS in 4m 14s","accounts_in_message":[],"_revision_number":2},{"id":"cd93cd10efcf60c48417abfbf38dddb779c7760d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-03 14:55:23.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d92ea1f8be490e5c47aed72dc7ed114372f0b9bd","revisions":{"945f8459adda21dbf41bfc8c6fed49adaf1c39f0":{"kind":"REWORK","_number":1,"created":"2019-06-18 14:29:29.000000000","uploader":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"ref":"refs/changes/97/665997/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/automaton","ref":"refs/changes/97/665997/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/automaton refs/changes/97/665997/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/automaton refs/changes/97/665997/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/automaton refs/changes/97/665997/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/automaton refs/changes/97/665997/1"}}},"commit":{"parents":[{"commit":"a9759e6b4d9f4c9138c43377f9340a5677996b84","subject":"Update to opendev","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/automaton/commit/a9759e6b4d9f4c9138c43377f9340a5677996b84"}]}],"author":{"name":"jacky06","email":"zhang.min@99cloud.net","date":"2019-06-18 14:29:09.000000000","tz":480},"committer":{"name":"jacky06","email":"zhang.min@99cloud.net","date":"2019-06-18 14:29:25.000000000","tz":480},"subject":"Sync Sphinx requirement","message":"Sync Sphinx requirement\n\n1. Sync sphinx dependency with global requirements. It caps python 2 since\nsphinx 2.0 no longer supports Python 2.7.\n2. Remove unncessary \"\u003d\u003d\"\n\nChange-Id: Ied4095b176b20795f6a8d4d7a84d01d958d76896\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/automaton/commit/945f8459adda21dbf41bfc8c6fed49adaf1c39f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/automaton/commit/945f8459adda21dbf41bfc8c6fed49adaf1c39f0"}]},"branch":"refs/heads/master"},"d92ea1f8be490e5c47aed72dc7ed114372f0b9bd":{"kind":"REWORK","_number":2,"created":"2019-06-18 16:57:58.000000000","uploader":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"ref":"refs/changes/97/665997/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/automaton","ref":"refs/changes/97/665997/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/automaton refs/changes/97/665997/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/automaton refs/changes/97/665997/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/automaton refs/changes/97/665997/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/automaton refs/changes/97/665997/2"}}},"commit":{"parents":[{"commit":"a9759e6b4d9f4c9138c43377f9340a5677996b84","subject":"Update to opendev","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/automaton/commit/a9759e6b4d9f4c9138c43377f9340a5677996b84"}]}],"author":{"name":"jacky06","email":"zhang.min@99cloud.net","date":"2019-06-18 14:29:09.000000000","tz":480},"committer":{"name":"jacky06","email":"zhang.min@99cloud.net","date":"2019-06-18 16:57:49.000000000","tz":480},"subject":"Sync Sphinx requirement","message":"Sync Sphinx requirement\n\n1. Sync sphinx dependency with global requirements. It caps python 2 since\nsphinx 2.0 no longer supports Python 2.7.\n2. Remove unncessary \"\u003d\u003d\"\n\nChange-Id: Ied4095b176b20795f6a8d4d7a84d01d958d76896\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/automaton/commit/d92ea1f8be490e5c47aed72dc7ed114372f0b9bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/automaton/commit/d92ea1f8be490e5c47aed72dc7ed114372f0b9bd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
