)]}'
{"id":"openstack%2Fautomaton~667679","triplet_id":"openstack%2Fautomaton~master~Ib9a720379ca997f848212eeca68ed9f32e4ed349","project":"openstack/automaton","branch":"master","topic":"bindep-fallback","hashtags":[],"change_id":"Ib9a720379ca997f848212eeca68ed9f32e4ed349","subject":"Add local bindep.txt","status":"MERGED","created":"2019-06-26 16:37:36.000000000","updated":"2019-07-01 07:43:43.000000000","submitted":"2019-07-01 07:43:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"667679-1561967023298-c2951fcc","meta_rev_id":"285c9ee56066868ed1fab83157f41f35c01aee64","_number":667679,"virtual_id_number":667679,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"value":0,"_account_id":6618,"name":"Ruby Loo","email":"opensrloo@gmail.com","username":"rloo"},{"value":2,"date":"2019-07-01 07:43:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},"all":[{"value":2,"date":"2019-07-01 05:52:26.000000000","_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"value":2,"date":"2019-06-28 23:24:14.000000000","_account_id":6618,"name":"Ruby Loo","email":"opensrloo@gmail.com","username":"rloo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},"all":[{"value":1,"date":"2019-07-01 05:52:26.000000000","_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"value":0,"_account_id":6618,"name":"Ruby Loo","email":"opensrloo@gmail.com","username":"rloo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"value":0,"_account_id":6618,"name":"Ruby Loo","email":"opensrloo@gmail.com","username":"rloo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6618,"name":"Ruby Loo","email":"opensrloo@gmail.com","username":"rloo"},{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-28 23:24:14.000000000","updated_by":{"_account_id":6618,"name":"Ruby Loo","email":"opensrloo@gmail.com","username":"rloo"},"reviewer":{"_account_id":6618,"name":"Ruby Loo","email":"opensrloo@gmail.com","username":"rloo"},"state":"REVIEWER"},{"updated":"2019-07-01 05:52:26.000000000","updated_by":{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},"reviewer":{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},"state":"REVIEWER"},{"updated":"2019-07-01 07:43:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5db5405f37fc2be1862446ef4c74e5ce07fb5431","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2019-06-26 16:37:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"18ff26e78a5605c8b75a92d3d3a57fe8dac1faa6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-26 18:47:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full http://logs.openstack.org/79/667679/1/check/tempest-full/908a232/ : SUCCESS in 2h 06m 53s\n- tempest-full-py3 http://logs.openstack.org/79/667679/1/check/tempest-full-py3/245b075/ : SUCCESS in 1h 55m 17s\n- openstack-tox-lower-constraints http://logs.openstack.org/79/667679/1/check/openstack-tox-lower-constraints/529d381/ : SUCCESS in 6m 00s\n- openstack-tox-pep8 http://logs.openstack.org/79/667679/1/check/openstack-tox-pep8/bd29fc9/ : SUCCESS in 3m 13s\n- openstack-tox-py27 http://logs.openstack.org/79/667679/1/check/openstack-tox-py27/6ebff7b/ : SUCCESS in 3m 14s\n- openstack-tox-py35 http://logs.openstack.org/79/667679/1/check/openstack-tox-py35/fd7234a/ : SUCCESS in 3m 27s\n- openstack-tox-py36 http://logs.openstack.org/79/667679/1/check/openstack-tox-py36/c10abdb/ : SUCCESS in 3m 06s\n- openstack-tox-py37 http://logs.openstack.org/79/667679/1/check/openstack-tox-py37/81ad9a0/ : SUCCESS in 6m 36s\n- openstack-tox-docs http://logs.openstack.org/79/667679/1/check/openstack-tox-docs/4b08288/html/ : SUCCESS in 4m 47s\n- build-openstack-releasenotes http://logs.openstack.org/79/667679/1/check/build-openstack-releasenotes/1a6ddb7/html/ : SUCCESS in 3m 05s","accounts_in_message":[],"_revision_number":1},{"id":"996eefe8947e33d22b0d9fa782f8ae0d44217bf6","author":{"_account_id":6618,"name":"Ruby Loo","email":"opensrloo@gmail.com","username":"rloo"},"date":"2019-06-28 23:24:14.000000000","message":"Patch Set 1: Code-Review+2\n\nThanks for fixing this!","accounts_in_message":[],"_revision_number":1},{"id":"acfa7eb76d7eea48381d5481e9fa1818b04adddd","author":{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},"date":"2019-07-01 05:52:26.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c1c4de6b960e6fa429b8bada11592e97bbdc3385","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-01 05:52:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3e74243bc616665b746abd94b9e6307454e71938","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-01 07:43:43.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"52003862a5fab8ffc700aaf6216f7c589e7311ea","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-01 07:43:43.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full http://logs.openstack.org/79/667679/1/gate/tempest-full/4ad9135/ : SUCCESS in 1h 42m 33s\n- tempest-full-py3 http://logs.openstack.org/79/667679/1/gate/tempest-full-py3/1a838c1/ : SUCCESS in 1h 46m 00s\n- openstack-tox-lower-constraints http://logs.openstack.org/79/667679/1/gate/openstack-tox-lower-constraints/fb467b0/ : SUCCESS in 4m 01s\n- openstack-tox-pep8 http://logs.openstack.org/79/667679/1/gate/openstack-tox-pep8/8bae0c6/ : SUCCESS in 2m 54s\n- openstack-tox-py27 http://logs.openstack.org/79/667679/1/gate/openstack-tox-py27/bf7827e/ : SUCCESS in 3m 17s\n- openstack-tox-py35 http://logs.openstack.org/79/667679/1/gate/openstack-tox-py35/ea77aa0/ : SUCCESS in 2m 59s\n- openstack-tox-py36 http://logs.openstack.org/79/667679/1/gate/openstack-tox-py36/5c4341a/ : SUCCESS in 3m 05s\n- openstack-tox-py37 http://logs.openstack.org/79/667679/1/gate/openstack-tox-py37/9332f6b/ : SUCCESS in 3m 44s\n- openstack-tox-docs http://logs.openstack.org/79/667679/1/gate/openstack-tox-docs/b7979f2/html/ : SUCCESS in 3m 04s\n- build-openstack-releasenotes http://logs.openstack.org/79/667679/1/gate/build-openstack-releasenotes/6b63a6f/html/ : SUCCESS in 3m 13s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"50f3de69ae2b00f9b6f95756eb072b44fae548d3","revisions":{"50f3de69ae2b00f9b6f95756eb072b44fae548d3":{"kind":"REWORK","_number":1,"created":"2019-06-26 16:37:36.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/79/667679/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/automaton","ref":"refs/changes/79/667679/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/automaton refs/changes/79/667679/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/automaton refs/changes/79/667679/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/automaton refs/changes/79/667679/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/automaton refs/changes/79/667679/1"}}},"commit":{"parents":[{"commit":"a9759e6b4d9f4c9138c43377f9340a5677996b84","subject":"Update to opendev","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/automaton/commit/a9759e6b4d9f4c9138c43377f9340a5677996b84"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2019-06-26 16:30:20.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2019-06-26 16:30:20.000000000","tz":120},"subject":"Add local bindep.txt","message":"Add local bindep.txt\n\nAs it was announced [1] global bindep-fallback.txt was removed and now\nprojects need to have a local bindep.txt to be able to install binary\ndependencies for testing.\n\nIn documentation zuul job graphviz package (\u0027dot\u0027 command) is needed.\nWithout that the job fails with:\n    dot command \u0027dot\u0027 cannot be run (needed for graphviz output),\n    check the graphviz_dot setting\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2019-June/007272.html\n\nChange-Id: Ib9a720379ca997f848212eeca68ed9f32e4ed349\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/automaton/commit/50f3de69ae2b00f9b6f95756eb072b44fae548d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/automaton/commit/50f3de69ae2b00f9b6f95756eb072b44fae548d3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
