)]}'
{"id":"openstack%2Fautomaton~668198","triplet_id":"openstack%2Fautomaton~stable%2Fstein~Ib9a720379ca997f848212eeca68ed9f32e4ed349","project":"openstack/automaton","branch":"stable/stein","topic":"bindep-fallback","hashtags":[],"change_id":"Ib9a720379ca997f848212eeca68ed9f32e4ed349","subject":"Add local bindep.txt","status":"MERGED","created":"2019-06-28 16:11:00.000000000","updated":"2019-07-08 04:47:09.000000000","submitted":"2019-07-08 04:47:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"668198-1562561229521-49c2a3a2","meta_rev_id":"8acf1e00dd731995401f16b6e06d034b86595e16","_number":668198,"virtual_id_number":668198,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"value":0,"date":"2019-06-28 17:00:01.000000000","_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2019-07-08 04:47:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},"all":[{"value":2,"date":"2019-07-08 03:07:41.000000000","_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":2,"date":"2019-07-04 10:51:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-07-04 22:32:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},"all":[{"value":1,"date":"2019-07-08 03:07:41.000000000","_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-04 10:51:32.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2019-07-04 22:32:12.000000000","updated_by":{"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"},"reviewer":{"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"},"state":"REVIEWER"},{"updated":"2019-07-08 03:07:41.000000000","updated_by":{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},"reviewer":{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},"state":"REVIEWER"},{"updated":"2019-07-08 04:47:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d35f0891e5d96bb10b577c0a63a2a5b51f36edd7","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2019-06-28 16:11:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"59d0319db6ac4bfd6b8e31431afd422d82d6d859","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2019-06-28 17:00:01.000000000","message":"Patch Set 1:\n\ntox-docs job fails [1] in stable periodic without this patch.\n\n[1] http://logs.openstack.org/periodic-stable/opendev.org/openstack/automaton/stable/stein/openstack-tox-docs/3e3c35d/job-output.txt.gz#_2019-06-28_06_27_17_423369","accounts_in_message":[],"_revision_number":1},{"id":"30f964fa19bd3bf72babd38de7cfb75294916d3d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-28 18:13:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full http://logs.openstack.org/98/668198/1/check/tempest-full/9fc290c/ : SUCCESS in 1h 43m 47s\n- tempest-full-py3 http://logs.openstack.org/98/668198/1/check/tempest-full-py3/c961b87/ : SUCCESS in 1h 22m 58s\n- openstack-tox-lower-constraints http://logs.openstack.org/98/668198/1/check/openstack-tox-lower-constraints/c807cc2/ : SUCCESS in 5m 48s\n- openstack-tox-pep8 http://logs.openstack.org/98/668198/1/check/openstack-tox-pep8/1fe70d2/ : SUCCESS in 3m 12s\n- openstack-tox-py27 http://logs.openstack.org/98/668198/1/check/openstack-tox-py27/90ec298/ : SUCCESS in 3m 17s\n- openstack-tox-py35 http://logs.openstack.org/98/668198/1/check/openstack-tox-py35/21b37c0/ : SUCCESS in 3m 13s\n- openstack-tox-py36 http://logs.openstack.org/98/668198/1/check/openstack-tox-py36/1c5035f/ : SUCCESS in 3m 22s\n- openstack-tox-docs http://logs.openstack.org/98/668198/1/check/openstack-tox-docs/9ff98d8/html/ : SUCCESS in 2m 57s\n- build-openstack-releasenotes http://logs.openstack.org/98/668198/1/check/build-openstack-releasenotes/8871577/html/ : SUCCESS in 3m 01s","accounts_in_message":[],"_revision_number":1},{"id":"024f5c413cae2d0a5252fdb9249c8301d312d715","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-07-04 10:51:32.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"9269d828fb2ac33cd522e552911410e7b2129f21","author":{"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"},"date":"2019-07-04 22:32:12.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"4da6626e16a0ba80de4618568528b8abc4c7f1dd","author":{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},"date":"2019-07-08 03:07:41.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2e1502a91cd302cba495661d3fba8fe6d992e531","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-08 03:07:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"875c6fa0f94511850e351663464545e44be83b94","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-08 04:47:09.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"06639c93fda6ae6daf227e800fb563cce9d3d271","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-08 04:47:09.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full http://logs.openstack.org/98/668198/1/gate/tempest-full/a4ed482/ : SUCCESS in 1h 37m 05s\n- tempest-full-py3 http://logs.openstack.org/98/668198/1/gate/tempest-full-py3/3faf553/ : SUCCESS in 1h 25m 25s\n- openstack-tox-lower-constraints http://logs.openstack.org/98/668198/1/gate/openstack-tox-lower-constraints/b5f2cee/ : SUCCESS in 5m 11s\n- openstack-tox-pep8 http://logs.openstack.org/98/668198/1/gate/openstack-tox-pep8/ba00617/ : SUCCESS in 3m 09s\n- openstack-tox-py27 http://logs.openstack.org/98/668198/1/gate/openstack-tox-py27/f36fb0e/ : SUCCESS in 4m 48s\n- openstack-tox-py35 http://logs.openstack.org/98/668198/1/gate/openstack-tox-py35/f6f957f/ : SUCCESS in 4m 02s\n- openstack-tox-py36 http://logs.openstack.org/98/668198/1/gate/openstack-tox-py36/9ee4843/ : SUCCESS in 4m 37s\n- openstack-tox-docs http://logs.openstack.org/98/668198/1/gate/openstack-tox-docs/639b1b5/html/ : SUCCESS in 3m 24s\n- build-openstack-releasenotes http://logs.openstack.org/98/668198/1/gate/build-openstack-releasenotes/6a1e779/html/ : SUCCESS in 4m 16s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"dba7395c2355edae19f0f73a3b45825b6e8b564c","revisions":{"dba7395c2355edae19f0f73a3b45825b6e8b564c":{"kind":"REWORK","_number":1,"created":"2019-06-28 16:11:00.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/98/668198/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/automaton","ref":"refs/changes/98/668198/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/automaton refs/changes/98/668198/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/automaton refs/changes/98/668198/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/automaton refs/changes/98/668198/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/automaton refs/changes/98/668198/1"}}},"commit":{"parents":[{"commit":"5b702b7e2c2c18cc0a7483f9b6870978f748d492","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/automaton/commit/5b702b7e2c2c18cc0a7483f9b6870978f748d492"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2019-06-26 16:30:20.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2019-06-28 16:10:22.000000000","tz":120},"subject":"Add local bindep.txt","message":"Add local bindep.txt\n\nAs it was announced [1] global bindep-fallback.txt was removed and now\nprojects need to have a local bindep.txt to be able to install binary\ndependencies for testing.\n\nIn documentation zuul job graphviz package (\u0027dot\u0027 command) is needed.\nWithout that the job fails with:\n    dot command \u0027dot\u0027 cannot be run (needed for graphviz output),\n    check the graphviz_dot setting\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2019-June/007272.html\n\nChange-Id: Ib9a720379ca997f848212eeca68ed9f32e4ed349\n(cherry picked from commit 50f3de69ae2b00f9b6f95756eb072b44fae548d3)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/automaton/commit/dba7395c2355edae19f0f73a3b45825b6e8b564c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/automaton/commit/dba7395c2355edae19f0f73a3b45825b6e8b564c"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
