)]}'
{"id":"openstack%2Fautomaton~668203","triplet_id":"openstack%2Fautomaton~stable%2Fqueens~Ib9a720379ca997f848212eeca68ed9f32e4ed349","project":"openstack/automaton","branch":"stable/queens","topic":"bindep-fallback","hashtags":[],"change_id":"Ib9a720379ca997f848212eeca68ed9f32e4ed349","subject":"Add local bindep.txt","status":"MERGED","created":"2019-06-28 16:46:26.000000000","updated":"2019-07-08 04:47:10.000000000","submitted":"2019-07-08 04:47:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"668203-1562561230102-b16e9fe7","meta_rev_id":"e0d047ba73e6bece3c95e25a6f07629c908dbc74","_number":668203,"virtual_id_number":668203,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"value":0,"date":"2019-06-28 17:02:03.000000000","_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2019-07-08 04:47:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},"all":[{"value":2,"date":"2019-07-08 03:09:12.000000000","_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":2,"date":"2019-07-04 10:51:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},"all":[{"value":1,"date":"2019-07-08 03:09:12.000000000","_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-04 10:51:27.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2019-07-08 03:09:12.000000000","updated_by":{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},"reviewer":{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},"state":"REVIEWER"},{"updated":"2019-07-08 04:47:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3d17afa2149a22fae9c65f07ef69c5b7b1375447","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2019-06-28 16:46:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ff0b96f68a78d8c143de6b2677324c4f681a5aab","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2019-06-28 17:02:03.000000000","message":"Patch Set 1:\n\ntox-docs job fails [1] in stable periodic without this patch.\n\n[1] http://logs.openstack.org/periodic-stable/opendev.org/openstack/automaton/stable/queens/build-openstack-sphinx-docs/55394be/job-output.txt.gz#_2019-06-28_06_21_44_816497","accounts_in_message":[],"_revision_number":1},{"id":"bba90a007e5a3de8d5794fb310fe57b7d6849ff8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-28 18:13:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/03/668203/1/check/openstack-tox-pep8/47514a9/ : SUCCESS in 2m 50s\n- openstack-tox-py27 http://logs.openstack.org/03/668203/1/check/openstack-tox-py27/a4d32da/ : SUCCESS in 2m 49s\n- openstack-tox-py35 http://logs.openstack.org/03/668203/1/check/openstack-tox-py35/ee146ad/ : SUCCESS in 3m 30s\n- openstack-tox-docs http://logs.openstack.org/03/668203/1/check/openstack-tox-docs/0861df8/html/ : SUCCESS in 2m 45s\n- tempest-full http://logs.openstack.org/03/668203/1/check/tempest-full/2b0d874/ : SUCCESS in 1h 23m 06s\n- build-openstack-releasenotes http://logs.openstack.org/03/668203/1/check/build-openstack-releasenotes/7b12116/html/ : SUCCESS in 3m 35s","accounts_in_message":[],"_revision_number":1},{"id":"d961f19321133fad02e5217c93972184954628a1","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-07-04 10:51:27.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"9978faf355f9350bf07e478ea81c2df20ee25f93","author":{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},"date":"2019-07-08 03:09:12.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4727169d829c38a41cfe3ad785a2c39725bf2a3c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-08 03:09:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d7dca4bae096fee6b5660c6498417bd0d03706a4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-08 04:47:09.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/03/668203/1/gate/openstack-tox-pep8/1c569eb/ : SUCCESS in 2m 37s\n- openstack-tox-py27 http://logs.openstack.org/03/668203/1/gate/openstack-tox-py27/56244e9/ : SUCCESS in 3m 56s\n- openstack-tox-py35 http://logs.openstack.org/03/668203/1/gate/openstack-tox-py35/c76ad53/ : SUCCESS in 2m 57s\n- openstack-tox-docs http://logs.openstack.org/03/668203/1/gate/openstack-tox-docs/a6200d3/html/ : SUCCESS in 3m 52s\n- tempest-full http://logs.openstack.org/03/668203/1/gate/tempest-full/b391dfb/ : SUCCESS in 1h 14m 53s\n- build-openstack-releasenotes http://logs.openstack.org/03/668203/1/gate/build-openstack-releasenotes/43e1285/html/ : SUCCESS in 4m 27s","accounts_in_message":[],"_revision_number":1},{"id":"e4b876371bf8a1938cebb1772ecd58f2abcd73ef","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-08 04:47:10.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"aabbec67b1e36fbe8ad9965c7aca23dfe46a9a56","revisions":{"aabbec67b1e36fbe8ad9965c7aca23dfe46a9a56":{"kind":"REWORK","_number":1,"created":"2019-06-28 16:46:26.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/03/668203/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/automaton","ref":"refs/changes/03/668203/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/automaton refs/changes/03/668203/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/automaton refs/changes/03/668203/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/automaton refs/changes/03/668203/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/automaton refs/changes/03/668203/1"}}},"commit":{"parents":[{"commit":"699f3d355272384564d999e1856beb6b1938c284","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/automaton/commit/699f3d355272384564d999e1856beb6b1938c284"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2019-06-26 16:30:20.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2019-06-28 16:26:53.000000000","tz":120},"subject":"Add local bindep.txt","message":"Add local bindep.txt\n\nAs it was announced [1] global bindep-fallback.txt was removed and now\nprojects need to have a local bindep.txt to be able to install binary\ndependencies for testing.\n\nIn documentation zuul job graphviz package (\u0027dot\u0027 command) is needed.\nWithout that the job fails with:\n    dot command \u0027dot\u0027 cannot be run (needed for graphviz output),\n    check the graphviz_dot setting\n\nConflicts:\n      tox.ini\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2019-June/007272.html\n\nChange-Id: Ib9a720379ca997f848212eeca68ed9f32e4ed349\n(cherry picked from commit 50f3de69ae2b00f9b6f95756eb072b44fae548d3)\n(cherry picked from commit dba7395c2355edae19f0f73a3b45825b6e8b564c)\n(cherry picked from commit 770f4b257253908da3f228197f87f0a36b701dd6)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/automaton/commit/aabbec67b1e36fbe8ad9965c7aca23dfe46a9a56"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/automaton/commit/aabbec67b1e36fbe8ad9965c7aca23dfe46a9a56"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
