)]}'
{"id":"openstack%2Fautomaton~668205","triplet_id":"openstack%2Fautomaton~stable%2Fpike~Ib9a720379ca997f848212eeca68ed9f32e4ed349","project":"openstack/automaton","branch":"stable/pike","topic":"bindep-fallback","hashtags":[],"change_id":"Ib9a720379ca997f848212eeca68ed9f32e4ed349","subject":"Add local bindep.txt","status":"MERGED","created":"2019-06-28 16:57:40.000000000","updated":"2019-07-08 12:31:53.000000000","submitted":"2019-07-08 12:31:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"668205-1562589113280-7de5f3e4","meta_rev_id":"92817021a342b71ca8b191ec0177e3c0c2fa0ffb","_number":668205,"virtual_id_number":668205,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"value":0,"date":"2019-07-08 08:27:43.000000000","_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2019-07-08 12:31:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},"all":[{"value":2,"date":"2019-07-08 03:09:20.000000000","_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":2,"date":"2019-07-04 10:51:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},"all":[{"value":1,"date":"2019-07-08 03:09:20.000000000","_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-04 10:51:18.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2019-07-08 03:09:20.000000000","updated_by":{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},"reviewer":{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},"state":"REVIEWER"},{"updated":"2019-07-08 12:31:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2bc9d090cbebc351ee3c078924e5ba6a7745db06","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2019-06-28 16:57:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d31ce7797c249d758531223ed164246c3208bb01","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2019-06-28 17:02:51.000000000","message":"Patch Set 1:\n\ntox-docs job fails [1] in stable periodic without this patch.\n\n[1] http://logs.openstack.org/periodic-stable/opendev.org/openstack/automaton/stable/pike/build-openstack-sphinx-docs/8d74967/","accounts_in_message":[],"_revision_number":1},{"id":"4566b1a9daca2837c320e3c9d9f1779487c4ee13","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-28 18:40:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/05/668205/1/check/openstack-tox-pep8/27df986/ : SUCCESS in 4m 02s\n- openstack-tox-py27 http://logs.openstack.org/05/668205/1/check/openstack-tox-py27/656e718/ : SUCCESS in 4m 14s\n- openstack-tox-py35 http://logs.openstack.org/05/668205/1/check/openstack-tox-py35/0171036/ : SUCCESS in 3m 54s\n- openstack-tox-docs http://logs.openstack.org/05/668205/1/check/openstack-tox-docs/1d8d4ed/html/ : SUCCESS in 3m 46s\n- tempest-full http://logs.openstack.org/05/668205/1/check/tempest-full/1f7d01a/ : SUCCESS in 1h 37m 37s\n- build-openstack-releasenotes http://logs.openstack.org/05/668205/1/check/build-openstack-releasenotes/943e2a3/html/ : SUCCESS in 2m 57s","accounts_in_message":[],"_revision_number":1},{"id":"83307a8749d420fcc20ee4dfa131ac0b59b7b742","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-07-04 10:51:18.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"7cdd286f99b53c5e5b5b41af5c38aeb58cf008cf","author":{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},"date":"2019-07-08 03:09:20.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ec00a6832fd63a60a3547fd7adadabf95740e11e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-08 03:09:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"35b1f97337d7fe7ae1db8757268c29647ffb5b81","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-08 04:47:10.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/05/668205/1/gate/openstack-tox-pep8/4b1beca/ : SUCCESS in 3m 29s\n- openstack-tox-py27 http://logs.openstack.org/05/668205/1/gate/openstack-tox-py27/fd35253/ : SUCCESS in 3m 31s\n- openstack-tox-py35 http://logs.openstack.org/05/668205/1/gate/openstack-tox-py35/2086306/ : SUCCESS in 3m 39s\n- openstack-tox-docs http://logs.openstack.org/05/668205/1/gate/openstack-tox-docs/0b1862b/html/ : SUCCESS in 4m 14s\n- tempest-full http://logs.openstack.org/05/668205/1/gate/tempest-full/1707d61/ : FAILURE in 1h 07m 15s\n- build-openstack-releasenotes http://logs.openstack.org/05/668205/1/gate/build-openstack-releasenotes/4558cf5/html/ : SUCCESS in 3m 22s","accounts_in_message":[],"_revision_number":1},{"id":"232e3fb34ee320679911bbfaa1f9b4a62b8603f3","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2019-07-08 08:27:43.000000000","message":"Patch Set 1:\n\nrecheck error_extend","accounts_in_message":[],"_revision_number":1},{"id":"9e659ef3ce518c7ff0e688a481d3ebe79ce0dabc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-08 10:56:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/05/668205/1/check/openstack-tox-pep8/d026dcd/ : SUCCESS in 4m 19s\n- openstack-tox-py27 http://logs.openstack.org/05/668205/1/check/openstack-tox-py27/9089391/ : SUCCESS in 4m 20s\n- openstack-tox-py35 http://logs.openstack.org/05/668205/1/check/openstack-tox-py35/165c00d/ : SUCCESS in 3m 38s\n- openstack-tox-docs http://logs.openstack.org/05/668205/1/check/openstack-tox-docs/eae7dc7/html/ : SUCCESS in 3m 15s\n- tempest-full http://logs.openstack.org/05/668205/1/check/tempest-full/3250656/ : SUCCESS in 1h 57m 22s\n- build-openstack-releasenotes http://logs.openstack.org/05/668205/1/check/build-openstack-releasenotes/45be1ac/html/ : SUCCESS in 3m 11s","accounts_in_message":[],"_revision_number":1},{"id":"686310ae4c8b62113c22c0a8f4962ce34363e2a7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-08 10:56:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"dbc1808ca43aed272edb2a13ab373ad222c4a6c0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-08 12:31:53.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"55bfd80511448782e7cc6e29ddf02ec235e02e3d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-08 12:31:53.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/05/668205/1/gate/openstack-tox-pep8/a79ccdf/ : SUCCESS in 3m 34s\n- openstack-tox-py27 http://logs.openstack.org/05/668205/1/gate/openstack-tox-py27/9beceb6/ : SUCCESS in 3m 04s\n- openstack-tox-py35 http://logs.openstack.org/05/668205/1/gate/openstack-tox-py35/0b87eec/ : SUCCESS in 3m 44s\n- openstack-tox-docs http://logs.openstack.org/05/668205/1/gate/openstack-tox-docs/8806b9b/html/ : SUCCESS in 3m 34s\n- tempest-full http://logs.openstack.org/05/668205/1/gate/tempest-full/24e1297/ : SUCCESS in 1h 33m 17s\n- build-openstack-releasenotes http://logs.openstack.org/05/668205/1/gate/build-openstack-releasenotes/3ffe296/html/ : SUCCESS in 3m 36s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5036b0779455a5432e6fc6a5939cbc3478c785f0","revisions":{"5036b0779455a5432e6fc6a5939cbc3478c785f0":{"kind":"REWORK","_number":1,"created":"2019-06-28 16:57:40.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/05/668205/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/automaton","ref":"refs/changes/05/668205/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/automaton refs/changes/05/668205/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/automaton refs/changes/05/668205/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/automaton refs/changes/05/668205/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/automaton refs/changes/05/668205/1"}}},"commit":{"parents":[{"commit":"4251470b4823b9cc7bfe65727a00230ce4dda7d9","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/automaton/commit/4251470b4823b9cc7bfe65727a00230ce4dda7d9"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2019-06-26 16:30:20.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2019-06-28 16:47:49.000000000","tz":120},"subject":"Add local bindep.txt","message":"Add local bindep.txt\n\nAs it was announced [1] global bindep-fallback.txt was removed and now\nprojects need to have a local bindep.txt to be able to install binary\ndependencies for testing.\n\nIn documentation zuul job graphviz package (\u0027dot\u0027 command) is needed.\nWithout that the job fails with:\n    dot command \u0027dot\u0027 cannot be run (needed for graphviz output),\n    check the graphviz_dot setting\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2019-June/007272.html\n\nChange-Id: Ib9a720379ca997f848212eeca68ed9f32e4ed349\n(cherry picked from commit 50f3de69ae2b00f9b6f95756eb072b44fae548d3)\n(cherry picked from commit dba7395c2355edae19f0f73a3b45825b6e8b564c)\n(cherry picked from commit 770f4b257253908da3f228197f87f0a36b701dd6)\n(cherry picked from commit aabbec67b1e36fbe8ad9965c7aca23dfe46a9a56)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/automaton/commit/5036b0779455a5432e6fc6a5939cbc3478c785f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/automaton/commit/5036b0779455a5432e6fc6a5939cbc3478c785f0"}]},"branch":"refs/heads/stable/pike"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
