)]}'
{"id":"openstack%2Fautomaton~690455","triplet_id":"openstack%2Fautomaton~master~Ia34b54872bded2edb015b2497a37f786e73e86bb","project":"openstack/automaton","branch":"master","hashtags":[],"change_id":"Ia34b54872bded2edb015b2497a37f786e73e86bb","subject":"Switch to Ussuri jobs","status":"MERGED","created":"2019-10-23 01:03:30.000000000","updated":"2019-12-02 14:47:38.000000000","submitted":"2019-12-02 14:42:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"690455-1575297774048-73dee3f6","meta_rev_id":"8fb9562c0eca237a305f3c7e9355a427bffb74a8","_number":690455,"virtual_id_number":690455,"owner":{"_account_id":23312,"name":"kangyufei","email":"kangyf@inspur.com","username":"kangyufei"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-02 14:42:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2019-12-02 13:06:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2019-10-28 13:35:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2019-12-02 13:06:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-28 13:35:45.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2019-12-02 13:06:41.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2019-12-02 14:42:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"38b70bb573ac0afb98b263b8fe7f540c51698649","author":{"_account_id":23312,"name":"kangyufei","email":"kangyf@inspur.com","username":"kangyufei"},"date":"2019-10-23 01:03:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6b84ec86eab26f0b115faee71405aaee6d706d03","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-23 02:48:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/53d6212ba47f463ca6e75e1e92b233c8 : SUCCESS in 1h 43m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/64118ce8f99c449d87be87d2f052f400 : SUCCESS in 1h 27m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/db59694cda564a5290fb2e58a80e4aed : SUCCESS in 4m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2386956f69c94f3a8b46234b56efc6e5 : SUCCESS in 3m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d3f706e4574b4d78ae8a77e63b7ffc3a : SUCCESS in 3m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9496d334898644fab79cbda2ceb48685 : SUCCESS in 3m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/193528050e4c413abbad685f35ae6e8a : SUCCESS in 3m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc4446d70ecb4e2c9d78afef8084047e : SUCCESS in 3m 36s","accounts_in_message":[],"_revision_number":1},{"id":"48afbd7a76a5603599d4ef1ac0ae659b8e4beaf7","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-10-28 13:35:45.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"62233ea985d23db04a1978a591feb17f722ce0a5","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-12-02 13:06:41.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"bda924ef02c1935c8016ca404f16e559dfa09c9d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-02 13:06:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7ec565ca71039d91d18160f26c0864135bdcebad","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-02 14:42:53.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/90e4c193ecdd46ee986d158cae1cc716 : SUCCESS in 1h 20m 48s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5219e1b524a74adeb8a813d6cfa7942a : SUCCESS in 1h 30m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ddcdf77389ef4532b1c6c9f0117f87a8 : SUCCESS in 4m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/39d3e5090f944a579bd7a918d36f9d2e : SUCCESS in 3m 18s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1b45bff66aa94340bad800fd521ae6cb : SUCCESS in 5m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/60f7ba7e3a674d58b9fb4ba372aaceda : SUCCESS in 3m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d90b2e1476bd43a190f9750a7dc9cf1a : SUCCESS in 3m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1b3b31829b5406293b537b48b393f2c : SUCCESS in 4m 00s","accounts_in_message":[],"_revision_number":1},{"id":"818595a6b2626a2f4c674d80e9f7475c9b2ff182","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-02 14:42:54.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"8fb9562c0eca237a305f3c7e9355a427bffb74a8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-02 14:47:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bbbe78e6360b49a5ae45c3d4d6e081fb : SUCCESS in 47s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2d4f5fbcfdb804c3107b80d810b398e064679aa0","revisions":{"2d4f5fbcfdb804c3107b80d810b398e064679aa0":{"kind":"REWORK","_number":1,"created":"2019-10-23 01:03:30.000000000","uploader":{"_account_id":23312,"name":"kangyufei","email":"kangyf@inspur.com","username":"kangyufei"},"ref":"refs/changes/55/690455/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/automaton","ref":"refs/changes/55/690455/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/automaton refs/changes/55/690455/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/automaton refs/changes/55/690455/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/automaton refs/changes/55/690455/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/automaton refs/changes/55/690455/1"}}},"commit":{"parents":[{"commit":"3d573d118d4c0c5c4dfbc4c2fd9b64f6a65a27b9","subject":"Blacklist sphinx 2.1.0 (autodoc bug)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/automaton/commit/3d573d118d4c0c5c4dfbc4c2fd9b64f6a65a27b9"}]}],"author":{"name":"kangyufei","email":"kangyf@inspur.com","date":"2019-10-23 01:03:16.000000000","tz":480},"committer":{"name":"kangyufei","email":"kangyf@inspur.com","date":"2019-10-23 01:03:26.000000000","tz":480},"subject":"Switch to Ussuri jobs","message":"Switch to Ussuri jobs\n\nChange-Id: Ia34b54872bded2edb015b2497a37f786e73e86bb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/automaton/commit/2d4f5fbcfdb804c3107b80d810b398e064679aa0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/automaton/commit/2d4f5fbcfdb804c3107b80d810b398e064679aa0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
