)]}'
{"id":"openstack%2Fautomaton~691622","triplet_id":"openstack%2Fautomaton~master~I9c6089c80aaa12b8f9380746e2caf519f9664bdf","project":"openstack/automaton","branch":"master","hashtags":[],"change_id":"I9c6089c80aaa12b8f9380746e2caf519f9664bdf","subject":"Stop testing python2.7","status":"ABANDONED","created":"2019-10-28 07:11:54.000000000","updated":"2019-12-18 15:06:05.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"7a14fa7764a4a5badfb5638319e572c1cb3f3661","_number":691622,"virtual_id_number":691622,"owner":{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":30619,"name":"Ajay Kumar","email":"mahajanajay33@gmail.com","username":"ajaykumar33"},{"tag":"autogenerated:zuul:check","value":1,"date":"2019-10-28 09:11:12.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":30619,"name":"Ajay Kumar","email":"mahajanajay33@gmail.com","username":"ajaykumar33"},"all":[{"value":-1,"date":"2019-10-28 07:26:15.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":-1,"date":"2019-10-31 10:31:01.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":30619,"name":"Ajay Kumar","email":"mahajanajay33@gmail.com","username":"ajaykumar33"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":30619,"name":"Ajay Kumar","email":"mahajanajay33@gmail.com","username":"ajaykumar33"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30619,"name":"Ajay Kumar","email":"mahajanajay33@gmail.com","username":"ajaykumar33"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30619,"name":"Ajay Kumar","email":"mahajanajay33@gmail.com","username":"ajaykumar33"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-28 07:26:15.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2019-10-28 09:11:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-10-31 10:31:01.000000000","updated_by":{"_account_id":30619,"name":"Ajay Kumar","email":"mahajanajay33@gmail.com","username":"ajaykumar33"},"reviewer":{"_account_id":30619,"name":"Ajay Kumar","email":"mahajanajay33@gmail.com","username":"ajaykumar33"},"state":"REVIEWER"}],"messages":[{"id":"832cbfa6caf105d339d6e5622b4645ff5bbfc1ea","author":{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},"date":"2019-10-28 07:11:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"90d655a8efb733bd5cc39a150412593a2aaf4501","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-10-28 07:26:15.000000000","message":"Patch Set 1: Code-Review-1\n\nstill runs py27.\n\nCommon libraries witched in phase 2, so WIP for now.\n\nhttp://lists.openstack.org/pipermail/openstack-discuss/2019-October/010356.html","accounts_in_message":[],"_revision_number":1},{"id":"cfd3ca59b9ab266eab625f6dbe9cadad8566bed6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-28 09:11:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/32d3e086726447a3b7c594a6aceab23c : SUCCESS in 3m 11s\n- tempest-full https://zuul.opendev.org/t/openstack/build/d009615d1b504b468f8fcf95bc37329a : SUCCESS in 1h 56m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d17d0561897b4595b9e99ab7842204fb : SUCCESS in 1h 31m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dd328cae5361472ea4dff3e46960ed3a : SUCCESS in 7m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5391dbe1fb241a5a0cb9a7c25d96291 : SUCCESS in 5m 00s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d0a4e8d51aed4294b9007344792ef82e : SUCCESS in 6m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/abd07c30aed242958e2ddbe2798687cb : SUCCESS in 6m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/183f4f63cacf4101830271aa6d9c8cb0 : SUCCESS in 6m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04095afee36b4c4c847c98ec2c3a52b9 : SUCCESS in 3m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a8ea5487eb7c4a619e89f2ab08a94901 : SUCCESS in 5m 40s","accounts_in_message":[],"_revision_number":1},{"id":"6e784a44104c4bc75d7601faa95465a549f6a120","author":{"_account_id":30619,"name":"Ajay Kumar","email":"mahajanajay33@gmail.com","username":"ajaykumar33"},"date":"2019-10-31 10:31:01.000000000","message":"Patch Set 1: Code-Review-1\n\nStill, zuul job for python 2 is running, as seen in the zuul check section of this patch.\n\nSo, please do these changes also:\n1) Remove the python2 job from zuul conf file.\n2) Please add the release note also.","accounts_in_message":[],"_revision_number":1},{"id":"7a14fa7764a4a5badfb5638319e572c1cb3f3661","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-12-18 15:06:05.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d3ff4bee4b65b0f1ca3bc0c837a64708809ea937","revisions":{"d3ff4bee4b65b0f1ca3bc0c837a64708809ea937":{"kind":"REWORK","_number":1,"created":"2019-10-28 07:11:54.000000000","uploader":{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},"ref":"refs/changes/22/691622/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/automaton","ref":"refs/changes/22/691622/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/automaton refs/changes/22/691622/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/automaton refs/changes/22/691622/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/automaton refs/changes/22/691622/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/automaton refs/changes/22/691622/1"}}},"commit":{"parents":[{"commit":"3d573d118d4c0c5c4dfbc4c2fd9b64f6a65a27b9","subject":"Blacklist sphinx 2.1.0 (autodoc bug)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/automaton/commit/3d573d118d4c0c5c4dfbc4c2fd9b64f6a65a27b9"}]}],"author":{"name":"zhangboye","email":"zhangboye@inspur.com","date":"2019-10-28 07:11:03.000000000","tz":480},"committer":{"name":"zhangboye","email":"zhangboye@inspur.com","date":"2019-10-28 07:11:40.000000000","tz":480},"subject":"Stop testing python2.7","message":"Stop testing python2.7\n\nChange-Id: I9c6089c80aaa12b8f9380746e2caf519f9664bdf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/automaton/commit/d3ff4bee4b65b0f1ca3bc0c837a64708809ea937"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/automaton/commit/d3ff4bee4b65b0f1ca3bc0c837a64708809ea937"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
