)]}'
{"id":"openstack%2Fautomaton~705616","triplet_id":"openstack%2Fautomaton~master~I42efda8d3fb5d300068e2fb18502fc266878e238","project":"openstack/automaton","branch":"master","topic":"reno-gitignore","hashtags":[],"change_id":"I42efda8d3fb5d300068e2fb18502fc266878e238","subject":"Ignore releasenote artifacts files.","status":"MERGED","created":"2020-02-04 09:07:16.000000000","updated":"2020-02-04 12:38:34.000000000","submitted":"2020-02-04 12:36:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"705616-1580819803124-6c82882c","meta_rev_id":"bf10e860ca9592bebca06085f616323a1d27baf0","_number":705616,"virtual_id_number":705616,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-04 12:36:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2020-02-04 09:38:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2020-02-04 09:38:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-04 09:38:34.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2020-02-04 12:36:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ce77a786213795113188100e799954e3d4bf138c","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-02-04 09:07:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"73b51c6a3a5cb09e000f927ab5cbdb8ef93733f3","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-02-04 09:13:53.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"cc2052bc4dd76a59af1c11500dc4309a6fc22c24","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-02-04 09:38:34.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nTrivial change","accounts_in_message":[],"_revision_number":2},{"id":"f1171a69bff01c71cf40673d0c7fdb576423658c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-04 11:06:55.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f82cf0bd441f451eb906f1b7a763ecce : SUCCESS in 1h 51m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/def8cec4292e421db7a05a2a1d917103 : SUCCESS in 4m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1cea7bb3c74c439fb6928d252264dabc : SUCCESS in 3m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/477d9afd52e74b78a9591869c9770b2b : SUCCESS in 4m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ab74b8efce5a4b8a8600f3dc65eea62d : SUCCESS in 4m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2a93d76142634732933be78e2025c64d : SUCCESS in 3m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8c6ed7359f14d8cb48df03b894f4db9 : SUCCESS in 3m 57s","accounts_in_message":[],"_revision_number":2},{"id":"5fd76b95830655b267befb884f7d8e99d5e29f6f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-04 11:08:35.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"974d9cafe91bd282db7f432f7a15c091e1a5095d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-04 12:36:42.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b4be8565afb74d47827155bb9998a7d9 : SUCCESS in 1h 27m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ff8ab5b9a9d2450cafeef845500d2f81 : SUCCESS in 3m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf9bb9e585364e3e9401cc3604d20d1a : SUCCESS in 4m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/49f5071e59564b0bb3ad22d134645501 : SUCCESS in 4m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e43b7b704531434c8b014909a35eeb3f : SUCCESS in 5m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc117e9f89ed4187b754f5b14b6eee19 : SUCCESS in 3m 50s","accounts_in_message":[],"_revision_number":2},{"id":"d3f596b5309e35f9bc3711048fee1d5bd1ec6294","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-04 12:36:43.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"d12fe953d18736c6a1ab016b48b9203fc9e066bf","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-04 12:38:34.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f83fe0192a64e4490ad7ea0a6dc8454 : SUCCESS in 1m 32s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"433c6d563c1c0395adfdf757740b37ed2cc068ea","revisions":{"9f45af976d405a52c7ea2ec2026491698e06b9b5":{"kind":"REWORK","_number":1,"created":"2020-02-04 09:07:16.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/16/705616/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/automaton","ref":"refs/changes/16/705616/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/automaton refs/changes/16/705616/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/automaton refs/changes/16/705616/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/automaton refs/changes/16/705616/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/automaton refs/changes/16/705616/1"}}},"commit":{"parents":[{"commit":"a1ed7110372e9aa69624396257917e3ffce3d73a","subject":"[ussuri][goal] Drop python 2.7 support and testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/automaton/commit/a1ed7110372e9aa69624396257917e3ffce3d73a"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-02-04 09:00:15.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-02-04 09:00:15.000000000","tz":60},"subject":"Ignore releasenote artifacts files.","message":"Ignore releasenote artifacts files.\n\nChange-Id: I42efda8d3fb5d300068e2fb18502fc266878e238\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/automaton/commit/9f45af976d405a52c7ea2ec2026491698e06b9b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/automaton/commit/9f45af976d405a52c7ea2ec2026491698e06b9b5"}]},"branch":"refs/heads/master"},"433c6d563c1c0395adfdf757740b37ed2cc068ea":{"kind":"REWORK","_number":2,"created":"2020-02-04 09:13:53.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/16/705616/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/automaton","ref":"refs/changes/16/705616/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/automaton refs/changes/16/705616/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/automaton refs/changes/16/705616/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/automaton refs/changes/16/705616/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/automaton refs/changes/16/705616/2"}}},"commit":{"parents":[{"commit":"a1ed7110372e9aa69624396257917e3ffce3d73a","subject":"[ussuri][goal] Drop python 2.7 support and testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/automaton/commit/a1ed7110372e9aa69624396257917e3ffce3d73a"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-02-04 09:00:15.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-02-04 09:13:23.000000000","tz":60},"subject":"Ignore releasenote artifacts files.","message":"Ignore releasenote artifacts files.\n\nChange-Id: I42efda8d3fb5d300068e2fb18502fc266878e238\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/automaton/commit/433c6d563c1c0395adfdf757740b37ed2cc068ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/automaton/commit/433c6d563c1c0395adfdf757740b37ed2cc068ea"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
