)]}'
{"id":"openstack%2Fautomaton~744294","triplet_id":"openstack%2Fautomaton~master~I711851a5494c9a47afee46060e65789bf12fffab","project":"openstack/automaton","branch":"master","topic":"migrate-to-focal","hashtags":[],"change_id":"I711851a5494c9a47afee46060e65789bf12fffab","subject":"DNM: testing gate on ubuntu focal","status":"ABANDONED","created":"2020-08-01 22:04:50.000000000","updated":"2020-08-02 23:26:48.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"29e77ac0e452069f9fa0f3a64e4f6344caece58d","_number":744294,"virtual_id_number":744294,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2020-08-02 22:40:37.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-02 22:40:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9e0df9fa028d8f2966aaf5c26c905da220cc3eb3","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-08-01 22:04:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f511f0af1b467b903c4148e30c80027c6fc4b123","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-01 23:21:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4657c460aa394f3a924944ea6d3f3fc5 : SUCCESS in 1h 15m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/900c8abb125e4222894f4bda7f2f08a2 : SUCCESS in 3m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bfee4faff3d44afaa4abb45967771edf : SUCCESS in 2m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/77714e114165445f922088fb3139b71f : SUCCESS in 4m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5844b4fa2753461ba8c87eb33ece7c18 : SUCCESS in 3m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6458fc1509d24242af859427db7d8fe4 : SUCCESS in 3m 45s","accounts_in_message":[],"_revision_number":1},{"id":"47c4df7982ed5273cd798931fc6b79ad037778cb","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-08-02 19:47:32.000000000","message":"Abandoned\n\nall working here, no updates needed for Focal migration.","accounts_in_message":[],"_revision_number":1},{"id":"f4ad8439b24987b1d131f3b65bfc2d1e95807e0a","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-08-02 20:04:04.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"e867383f2c5f93015ffd06f9958206438f4d6cfb","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-08-02 20:04:16.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"7dd24a6faa8c5a35892dbf7e599103893ec1ea97","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-02 22:40:37.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/88c2c09553144e82a19ec8616326f7f8 : SUCCESS in 1h 21m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9c5aa77ea33442efb6cf662f199be1da : SUCCESS in 4m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f3b840dfcb4341c3aff091399be43220 : SUCCESS in 4m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dabb81b8a46a459c8b07184e1b19c842 : SUCCESS in 5m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b088fea7a39f4dcea6d97ff284ecc72f : SUCCESS in 3m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70674d3a3e7243119579f7df5be49579 : SUCCESS in 4m 00s","accounts_in_message":[],"_revision_number":2},{"id":"902bfcce958fd1ff6fb310a088883560bd7112d3","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-08-02 23:26:48.000000000","message":"Abandoned\n\nall working here, no updates needed for Focal migration.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"91ef8722ec23223999eefccbda980fd567b06aed","revisions":{"62e9d133f9aa9daf6040b9820a39d63295ab1328":{"kind":"REWORK","_number":1,"created":"2020-08-01 22:04:50.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/94/744294/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/automaton","ref":"refs/changes/94/744294/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/automaton refs/changes/94/744294/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/automaton refs/changes/94/744294/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/automaton refs/changes/94/744294/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/automaton refs/changes/94/744294/1"}}},"commit":{"parents":[{"commit":"a30803c50ee8551e8e3d168c318a0c49559e0899","subject":"Update lower-constraints list","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/automaton/commit/a30803c50ee8551e8e3d168c318a0c49559e0899"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-08-01 22:03:43.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-08-01 22:04:51.000000000","tz":-300},"subject":"DNM: testing gate on ubuntu focal","message":"DNM: testing gate on ubuntu focal\n\nThis tests:\n- unit tests jobs\n- functional integration tests jobs\n- docs jobs\nStory: #2007865\nTask: #40207\n\nChange-Id: I711851a5494c9a47afee46060e65789bf12fffab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/automaton/commit/62e9d133f9aa9daf6040b9820a39d63295ab1328"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/automaton/commit/62e9d133f9aa9daf6040b9820a39d63295ab1328"}]},"branch":"refs/heads/master"},"91ef8722ec23223999eefccbda980fd567b06aed":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-08-02 20:04:16.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/94/744294/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/automaton","ref":"refs/changes/94/744294/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/automaton refs/changes/94/744294/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/automaton refs/changes/94/744294/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/automaton refs/changes/94/744294/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/automaton refs/changes/94/744294/2"}}},"commit":{"parents":[{"commit":"a30803c50ee8551e8e3d168c318a0c49559e0899","subject":"Update lower-constraints list","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/automaton/commit/a30803c50ee8551e8e3d168c318a0c49559e0899"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-08-01 22:03:43.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-08-02 20:04:14.000000000","tz":0},"subject":"DNM: testing gate on ubuntu focal","message":"DNM: testing gate on ubuntu focal\n\nThis tests:\n- unit tests jobs\n- functional integration tests jobs\n- docs jobs\nStory: #2007865\nTask: #40207\n\nDepends-On: https://review.opendev.org/#/c/734700/\n\nChange-Id: I711851a5494c9a47afee46060e65789bf12fffab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/automaton/commit/91ef8722ec23223999eefccbda980fd567b06aed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/automaton/commit/91ef8722ec23223999eefccbda980fd567b06aed"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
